)]}'
{"id":"openstack%2Fkeystone~903088","triplet_id":"openstack%2Fkeystone~master~I361b2d0df03af4fcd5b3a477b03c6b056acb9096","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2023-12-07 16:49:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I361b2d0df03af4fcd5b3a477b03c6b056acb9096","subject":"[DNM] Gate health check","status":"ABANDONED","created":"2023-12-07 15:01:23.000000000","updated":"2023-12-07 16:49:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"31a5267327b779ca981a6f185117793ca47b3836","_number":903088,"virtual_id_number":903088,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-12-07 16:29:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-07 15:14:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-07 16:29:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"85a53b21c55a449d3e43f242baf57e0377bb603d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2023-12-07 15:01:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"610aa472eda8339a368ecd2b1e2f90ee1f3382d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 15:14:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/437d5b6328724cba81eaea04d20c6f5a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a77c772e3b74d40ae58948d12a616a2 : SUCCESS in 11m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4f7ce28a788d4ed7a7ef2813ba4fddfd : SUCCESS in 12m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3e79d620c2007d5ab9d8449c3ff1cec6f6a496a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 16:29:54.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6d8a85552dc4aaf88099881804208ab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb5b1cc598134320b21e54b35a10c78e : SUCCESS in 14m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f56d0f0bf3d4c579f87cd74489babf8 : FAILURE in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/206cd9cc553a4d3ba36dddbf34109e3b : SUCCESS in 11m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b044104e007a41de8f0e6fc536e435f0 : SUCCESS in 12m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e61cde48408245e6b1014b1cc680da1e : SUCCESS in 14m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/59e96ba5d21645e0928bfdcb901e2bff : SUCCESS in 59m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/32e65a46720e4e9b980a7ac825f50161 : SUCCESS in 1h 20m 07s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5b22b74e66fe4fa79cc26f532cb0b92c : SUCCESS in 14m 56s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/99037236c120480e8e4f6e75928d9881 : SUCCESS in 40m 56s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/a56df264af924f9dad860130a67030da : SUCCESS in 41m 21s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbffdf7e36a642eaadf56766ba08d785 : SUCCESS in 37m 07s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/62ba780f215147a78e464f55da1ab198 : SUCCESS in 27m 18s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/a97538ea3ffc448f858993f5cc359f38 : SUCCESS in 22m 54s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c62245b6f018475b923bb355bd30d8f8 : FAILURE in 26m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d570b995f5445a4aa9241a74befd35d : SUCCESS in 1h 09m 36s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/18991fef38be4befa233d1b482d8b2a0 : SUCCESS in 53m 30s","accounts_in_message":[],"_revision_number":1},{"id":"31a5267327b779ca981a6f185117793ca47b3836","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2023-12-07 16:49:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6ab383a41dcd80c0843476506c5522c0476b52a","revisions":{"a6ab383a41dcd80c0843476506c5522c0476b52a":{"kind":"REWORK","_number":1,"created":"2023-12-07 15:01:23.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/88/903088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/88/903088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/903088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/903088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/903088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/88/903088/1"}}},"commit":{"parents":[{"commit":"262d763f79a9b97996ec51dca3399994b97bf4e0","subject":"Merge \"Remove unnecessary shebang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/262d763f79a9b97996ec51dca3399994b97bf4e0"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2023-12-07 15:00:45.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2023-12-07 15:00:45.000000000","tz":-300},"subject":"[DNM] Gate health check","message":"[DNM] Gate health check\n\nChange-Id: I361b2d0df03af4fcd5b3a477b03c6b056acb9096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a6ab383a41dcd80c0843476506c5522c0476b52a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a6ab383a41dcd80c0843476506c5522c0476b52a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
