)]}'
{"id":"openstack%2Fkeystone~904305","triplet_id":"openstack%2Fkeystone~master~Iee22be887130dd171ae8038f5ed3bb365e2b3ade","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2024-01-19 17:03:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iee22be887130dd171ae8038f5ed3bb365e2b3ade","subject":"python 3.12: use raw string","status":"MERGED","created":"2023-12-23 12:02:40.000000000","updated":"2024-01-19 17:04:36.000000000","submitted":"2024-01-19 17:03:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904305","meta_rev_id":"90edfb2fece7d5982cd32444e3934a6ad5c54828","_number":904305,"virtual_id_number":904305,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-19 17:03:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2024-01-05 15:44:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2024-01-19 15:05:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2024-01-19 15:05:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-23 12:12:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-23 13:19:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-05 15:44:23.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2024-01-19 15:05:18.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"51690d69f710567969f9cb3991c7854ede66ca0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2023-12-23 12:02:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1428644d7124fca8bed72645cad52640a015a73d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-23 12:12:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d7844f0105c4d658e167e7fa8f8ff3f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4bedac2325bc475ab24d04fbd5d6f565 : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f2703eeab484f13a9a610a07b81f629 : SUCCESS in 9m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a63f160e2ce9f6dc3ff62ceb5c0d9df12ee3807","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-23 13:19:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1b24bf45f6d4b408b5e7890f1a8dd76\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/606b967a5fa8438abed80ff5cf3af1ed : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20fa67fa07da4af7b2208607e7504472 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2b95acf01be4d548a64344f153f6c56 : SUCCESS in 9m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c4293695330b4b828b70e09ec6c80dfc : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92b9511af4f44279be5af0905b3d7299 : SUCCESS in 10m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/39c5cdb849aa47cba946cd48eae24711 : SUCCESS in 40m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad6ab0f87c334ef281d24948f3386b9b : SUCCESS in 59m 43s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4c135b2897434a18bc3ca1452b8d2f48 : SUCCESS in 10m 58s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/fd36233c0566459dafd9ff1248536939 : SUCCESS in 19m 40s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/1ac9c76f08bd482b9df1495ae7d0b99a : SUCCESS in 26m 48s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4a1193c627c444b87671e81b9b227b8 : SUCCESS in 21m 27s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/384d56c510514069b8f4bdc6a9decbd6 : SUCCESS in 26m 19s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/68c1815ae7384a948f0b2cab3d2abe0a : SUCCESS in 15m 03s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/d4f9a5a265374aa6a0c73ee4b1fdeb03 : FAILURE in 23m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a5c8a0e656948fb9bf1baef57e09399 : SUCCESS in 46m 26s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/216c00d2dc7546be96775164bbccb7ca : SUCCESS in 47m 20s","accounts_in_message":[],"_revision_number":1},{"id":"df7591660eac3ab984e4a19c5353d34262be576e","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-01-05 15:44:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ee7ea4db3c0fe4a0b419c484ebf956a21a04da68","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-01-19 15:05:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e93a9d658b701036110fe2eb6081e5dd2870f9a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 15:05:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b1dc3695df2ccdb991bef9eaa42773687b52eee4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 17:03:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14e39e54f3384cceaad99ba3188c8f40\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/679026fdcebc4148bbbb2f97e0c4baca : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/baa3bc8a6df345ff82f89effd0fc9584 : SUCCESS in 10m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4801b8fcdba34adb8c63723ace491953 : SUCCESS in 13m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74a88814ed094a19856e73d82c1df82d : SUCCESS in 11m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/a5d68477575a468ba6062cfbb29e3acd : SUCCESS in 1h 02m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ebd841cd814448e59890c60ecfd4a14d : SUCCESS in 1h 51m 39s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a6758fdafe534e9d8394bb03d8ccd73d : SUCCESS in 12m 34s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/157e58dc06074bdea5d934ad2a34845e : SUCCESS in 29m 54s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/610ea22f6a8e4f1783f75621f3e833b4 : SUCCESS in 34m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d875bfc9b19742f7b74ee655e10f18a7 : SUCCESS in 45m 17s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/0a413ad781a44627bcb52d58c3da03c9 : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6392a7bfd54874efb531ca2320a4fdfcdc0d5858","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 17:03:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"90edfb2fece7d5982cd32444e3934a6ad5c54828","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 17:04:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4c7f89c89a6405e808420c7930b0df2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d02548dba2a44994ba0c120816e67034 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"45dde73ba3a7ec45d3556ee670699e11638a6ff0","revisions":{"45dde73ba3a7ec45d3556ee670699e11638a6ff0":{"kind":"REWORK","_number":1,"created":"2023-12-23 12:02:40.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/05/904305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/05/904305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/904305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/904305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/904305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/05/904305/1"}}},"commit":{"parents":[{"commit":"9e4a3157dd326801b9ae0ef2b7934c982259b3dd","subject":"Merge \"Fix typo in cmd/status.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9e4a3157dd326801b9ae0ef2b7934c982259b3dd"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2023-12-23 12:02:00.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2023-12-23 12:02:00.000000000","tz":60},"subject":"python 3.12: use raw string","message":"python 3.12: use raw string\n\nThis fixes \"SyntaxWarning: invalid escape sequence \u0027\\d\u0027\nwhen installing python3-keystone.\n\nChange-Id: Iee22be887130dd171ae8038f5ed3bb365e2b3ade\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/45dde73ba3a7ec45d3556ee670699e11638a6ff0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/45dde73ba3a7ec45d3556ee670699e11638a6ff0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
