)]}'
{"id":"openstack%2Fkeystone~906312","triplet_id":"openstack%2Fkeystone~master~I5307464684c8e2e979248cb6cf9d3ada0257ad73","project":"openstack/keystone","branch":"master","topic":"python-3.12","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-22 21:14:13.000000000","reason":"Change was abandoned"},"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2024-02-22 21:14:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5307464684c8e2e979248cb6cf9d3ada0257ad73","subject":"Fix datetime usage under Python 3.12","status":"ABANDONED","created":"2024-01-22 18:21:29.000000000","updated":"2024-02-23 14:41:56.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"a11fcc676a1a1b579a56c7549d279522fd1af28a","_number":906312,"virtual_id_number":906312,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-01-23 15:23:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"all":[{"value":-1,"date":"2024-02-05 10:23:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}],"REVIEWER":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-22 18:30:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-22 19:37:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-23 08:35:02.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-02-05 10:23:23.000000000","updated_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"REVIEWER"},{"updated":"2024-02-23 14:41:56.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"}],"messages":[{"id":"2df55f4e1f33d7aa51661683c24fc92c34daaa79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-01-22 18:21:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"933c6959aa81364cf6c20b2bcfdbfe53ced56a56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 18:30:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b12fd0c27d8c488bbfb14ed55c0a3be7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3feddf8123b48bbb62c08e5df82a108 : FAILURE in 8m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4f3f971018184676af0be7a947380acb : FAILURE in 8m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5348c98e10f8d6b36deed8266a68bcc709b17b44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 19:37:21.000000000","message":"Patch Set 1: Verified-1\n\n(13 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a77c57d5f5f43fbb14896015bb8a0fb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87d1a50994204dffa572e1de9e20662a : FAILURE in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38f6c2f5a4e34848a343b8b12d178e2f : FAILURE in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce9206d755f242498607f524e007f5cc : FAILURE in 11m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/056da94055e648d685a22daae3d36e07 : SUCCESS in 14m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbda671d6e9240f380b26f8b1904eaf6 : FAILURE in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/89389d045b2542c68734a5bf74520c06 : FAILURE in 1h 05m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba0d2dc076e84ca085aa157aa86fa6f0 : FAILURE in 14m 09s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7fedfa38b508426d89ddddd436d0da57 : FAILURE in 12m 05s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/06f5ea987ddc4be19b1760b1fa6e1aae : FAILURE in 17m 34s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/b6764c6c07544d3281c47ba4ebad8661 : FAILURE in 26m 25s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/986a611d14944ca595b75622050fe5e3 : FAILURE in 17m 15s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/7cd205c4c8d84171893de2d906288b6f : FAILURE in 26m 04s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/de844a9cb5784498b66aca34a772c066 : POST_FAILURE in 19m 28s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/474a5d94b4c540f7a70f99d4b0f05916 : FAILURE in 12m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa67d19233ff45eb90a296c9ef1e69f8 : FAILURE in 27m 32s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/460e3d6709a04755938f493e8794b6ba : FAILURE in 14m 38s","accounts_in_message":[],"_revision_number":1},{"id":"10d69b1fffa6ecd8ed9547a3df8cef3d953ae8fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-01-23 08:31:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f18d4d9992e20eb2fe9a7b7dc53f56d94e21eb82","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-23 08:35:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"684f7df1bf06c8161efcc0eaed69cf50121a75bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 08:40:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd89ece37a8841f5874d81e553b62478\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/388bb85c19a442358f7e33fc4c57cdee : FAILURE in 8m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b3425148d5f499c961100aae78ed386 : FAILURE in 8m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d6c6dc43fac6d2dd205c55d4320f70b305791358","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 09:45:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6118a55634184dadbf69cf3ced5d3619\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aea37e17b52d4b6f8decd24f2956b7f1 : FAILURE in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d630a0c4e8d47578c80bc684e5b0e60 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51271396478f4c439ab115bc7ae57a33 : FAILURE in 10m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c54615776700414cb9cd21cc557e7822 : FAILURE in 12m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5f046e8b0b24fce89bedf036af8cc53 : SUCCESS in 12m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/3ebbce789531427c8069e9f6e06ab749 : FAILURE in 1h 05m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e838844bb3ce4c8d80e3be8544b73db7 : FAILURE in 28m 05s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1f699d54da4849a6a6083f3bd907b3cd : FAILURE in 11m 53s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ad779f03a8d34ed2b0c4cc94e88713dd : FAILURE in 28m 10s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/bc17262c550e4a408fadcbf72dc1403e : FAILURE in 23m 39s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8251fbb49414f8da9c6426f7e8e2812 : FAILURE in 11m 16s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/c2c2769ae52441c58fee6bb9308d0ea9 : FAILURE in 23m 00s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/3d47a6fc3c704f7db7287cb661e0642c : POST_FAILURE in 6m 55s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/eee1091db9e446b3af9604c566e1cd54 : FAILURE in 25m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b20200f8cd5f49dc8b7cc5a1eefeb3ec : FAILURE in 24m 56s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/fb172ce9206149f19dfecacf0228dd8e : FAILURE in 21m 20s","accounts_in_message":[],"_revision_number":2},{"id":"2b3594cfe994f22af90229bdb2e64a454318b28a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-01-23 14:17:59.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"28cc9f8caded9eecae81eebc233ef10dea59bbba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-01-23 14:18:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9a56a254d745e34f9e0e7b131b714a65731dc2e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 14:57:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9838910a70041028c9f29026e70dbbe\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d130769239d42e4976b53cbcc84e710 : FAILURE in 36m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/428845b7ccb74e92957c5613a5a51e3c : FAILURE in 13m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d023a040268044fd3bb153f164fe9a8521ea2094","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 15:23:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c2462753ae0450d83fd18abfc9afe40\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cb189005c8e439d926c85195ea8ba8a : FAILURE in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d12f226c89fc44569dc85eea09f2ea19 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ce85cdec9cb4e56bc5ee9f4e2c13cb6 : FAILURE in 11m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5fd65e6ab2724633926a93593492738a : SUCCESS in 13m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07a00bf21557445d94c5601e624224d8 : SUCCESS in 13m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/a192cfaacfd04ee5a42742b40dbb43f6 : FAILURE in 58m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96def906ca434c69b8295c0e1cc69a37 : FAILURE in 13m 04s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f73909443a8c4407a137239ece2feee5 : FAILURE in 12m 30s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ec3c72ace23f457a8badb99a6fbbb041 : FAILURE in 23m 59s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/dc12d337ca6f4ef3921c271d72c5b432 : FAILURE in 24m 48s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74de47b38e2947b788393f06e3eef251 : FAILURE in 25m 12s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/3fa6f0b85d3c4992b8bb9b40689c307a : FAILURE in 27m 56s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/12f39b637647439c8023d5dae8c76093 : POST_FAILURE in 10m 03s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/a3ecd5037aef4d099f6a0e244cef6abb : FAILURE in 12m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/666e917e241f42cc969788fac389bdd2 : FAILURE in 25m 40s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/1825e266b3e8421ab89f6f92bba55a69 : FAILURE in 11m 53s","accounts_in_message":[],"_revision_number":4},{"id":"0ab6bf25430cb98d19a128b8d15dba73671663ca","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-02-05 10:23:23.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e0276298344348f7b06a8f5e7a63f0c2bbbeeac8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-22 21:14:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fc8c61fc544f841dd835d11992861258a13a2f93","revisions":{"4b19406630891a9c98310c9c2376b96107ef617f":{"kind":"REWORK","_number":1,"created":"2024-01-22 18:21:29.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/12/906312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/12/906312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/906312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/906312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/906312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/12/906312/1"}}},"commit":{"parents":[{"commit":"bb4b7abc8ddeb6a3c8d5417e9b66c79e29f376f6","subject":"Merge \"python 3.12: use raw string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bb4b7abc8ddeb6a3c8d5417e9b66c79e29f376f6"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-01-22 18:19:16.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-01-22 18:21:26.000000000","tz":60},"subject":"Fix datetime usage under Python 3.12","message":"Fix datetime usage under Python 3.12\n\nAs per the Python 3.12 release notes, a number of functions were\ndeprecated: https://docs.python.org/3/whatsnew/3.12.html\n\n\"datetime.datetime’s utcnow() and utcfromtimestamp() are deprecated\"\n\nThis broke 2000+ unit tests when running them under Python 3.12.\nThis patch fixes all of this.\n\nChange-Id: I5307464684c8e2e979248cb6cf9d3ada0257ad73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4b19406630891a9c98310c9c2376b96107ef617f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4b19406630891a9c98310c9c2376b96107ef617f"}]},"branch":"refs/heads/master"},"e956a59f3e8be0fa5123cfc2b585f401856200d2":{"kind":"REWORK","_number":2,"created":"2024-01-23 08:31:26.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/12/906312/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/12/906312/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/906312/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/906312/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/906312/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/12/906312/2"}}},"commit":{"parents":[{"commit":"bb4b7abc8ddeb6a3c8d5417e9b66c79e29f376f6","subject":"Merge \"python 3.12: use raw string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bb4b7abc8ddeb6a3c8d5417e9b66c79e29f376f6"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-01-22 18:19:16.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-01-23 08:31:17.000000000","tz":60},"subject":"Fix datetime usage under Python 3.12","message":"Fix datetime usage under Python 3.12\n\nAs per the Python 3.12 release notes, a number of functions were\ndeprecated: https://docs.python.org/3/whatsnew/3.12.html\n\n\"datetime.datetime’s utcnow() and utcfromtimestamp() are deprecated\"\n\nThis broke 2000+ unit tests when running them under Python 3.12.\nThis patch fixes all of this.\n\nChange-Id: I5307464684c8e2e979248cb6cf9d3ada0257ad73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e956a59f3e8be0fa5123cfc2b585f401856200d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e956a59f3e8be0fa5123cfc2b585f401856200d2"}]},"branch":"refs/heads/master"},"056f702ecd2fb6c6dfbc55f9a6effeb0a350bc39":{"kind":"REWORK","_number":3,"created":"2024-01-23 14:17:59.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/12/906312/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/12/906312/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/906312/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/906312/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/906312/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/12/906312/3"}}},"commit":{"parents":[{"commit":"bb4b7abc8ddeb6a3c8d5417e9b66c79e29f376f6","subject":"Merge \"python 3.12: use raw string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bb4b7abc8ddeb6a3c8d5417e9b66c79e29f376f6"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-01-22 18:19:16.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-01-23 14:17:50.000000000","tz":60},"subject":"Fix datetime usage under Python 3.12","message":"Fix datetime usage under Python 3.12\n\nAs per the Python 3.12 release notes, a number of functions were\ndeprecated: https://docs.python.org/3/whatsnew/3.12.html\n\n\"datetime.datetime’s utcnow() and utcfromtimestamp() are deprecated\"\n\nThis broke 2000+ unit tests when running them under Python 3.12.\nThis patch fixes all of this.\n\nChange-Id: I5307464684c8e2e979248cb6cf9d3ada0257ad73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/056f702ecd2fb6c6dfbc55f9a6effeb0a350bc39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/056f702ecd2fb6c6dfbc55f9a6effeb0a350bc39"}]},"branch":"refs/heads/master"},"fc8c61fc544f841dd835d11992861258a13a2f93":{"kind":"REWORK","_number":4,"created":"2024-01-23 14:18:47.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/12/906312/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/12/906312/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/906312/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/906312/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/906312/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/12/906312/4"}}},"commit":{"parents":[{"commit":"bb4b7abc8ddeb6a3c8d5417e9b66c79e29f376f6","subject":"Merge \"python 3.12: use raw string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bb4b7abc8ddeb6a3c8d5417e9b66c79e29f376f6"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-01-22 18:19:16.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-01-23 14:18:39.000000000","tz":60},"subject":"Fix datetime usage under Python 3.12","message":"Fix datetime usage under Python 3.12\n\nAs per the Python 3.12 release notes, a number of functions were\ndeprecated: https://docs.python.org/3/whatsnew/3.12.html\n\n\"datetime.datetime’s utcnow() and utcfromtimestamp() are deprecated\"\n\nThis broke 2000+ unit tests when running them under Python 3.12.\nThis patch fixes all of this.\n\nChange-Id: I5307464684c8e2e979248cb6cf9d3ada0257ad73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fc8c61fc544f841dd835d11992861258a13a2f93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fc8c61fc544f841dd835d11992861258a13a2f93"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
