)]}'
{"id":"openstack%2Fkeystone~906953","triplet_id":"openstack%2Fkeystone~master~If78d0b93665410b86754ea35653ca9d4c15c81c5","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-03-08 15:27:32.000000000","reason":"\u003cGERRIT_ACCOUNT_7973\u003e replied on the change","reason_account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2024-03-12 14:42:59.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-12 17:22:44.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2024-03-12 17:22:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If78d0b93665410b86754ea35653ca9d4c15c81c5","subject":"Drop remaining references to eventlet options","status":"MERGED","created":"2024-01-27 12:03:16.000000000","updated":"2024-03-13 13:04:23.000000000","submitted":"2024-03-12 17:22:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906953","meta_rev_id":"ff218f25315f20189bc6b27148cf5cafb99f1656","_number":906953,"virtual_id_number":906953,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-12 17:22:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2024-03-08 15:27:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2024-03-12 14:42:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2024-03-12 14:42:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-27 12:13:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-27 13:42:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-07 07:20:59.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-03-07 07:20:59.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2024-03-07 07:20:59.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"94c49116d117ddb35e2e541ae263ba9d82a29c06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-27 12:03:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d5c95621e2451e4eb3a038e8a5770db8e24e76c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-27 12:13:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85ed51ed35d947d68d03fbe9be727265\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d65a95c5ec224a81bbf2d7290842cc61 : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/571ea89244b4442581f09e9adf179db4 : SUCCESS in 9m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63c3d462ffb947898c41633934223742195a2f4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-27 13:42:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/512657e331624a589d09055df8b17a49\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/874ff12cc7e147378450d4e7c3e6094c : SUCCESS in 14m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20a824b39ac149229db48e361ea1fe8c : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5be58a6524a4079a29f31550a89fc2f : SUCCESS in 13m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c41bf592ced44c8baeb93c5e3c743908 : SUCCESS in 11m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cae2337ff4e046b9849a14b69506ada9 : SUCCESS in 13m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/5ad6daddec834548bc40ee4b7eab9f82 : SUCCESS in 1h 08m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c39f9d07ea1a48b49f45c0209f11654c : SUCCESS in 1h 33m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43f039b4a3674b68979fecb4a0a13933 : SUCCESS in 11m 36s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1f6d77b1878a4b6cb8dea2deae082c43 : SUCCESS in 11m 15s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/594fbfe7f3254ea3b7cbcd3f3f883da2 : SUCCESS in 18m 19s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/644846061a04441a8cf4f5eb1a1973f2 : SUCCESS in 24m 01s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc92a64a97c04497afd05d19dd52fd6c : FAILURE in 19m 01s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/85d232f0efe1415d992bb4b52c1bc427 : SUCCESS in 17m 38s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/2ba6d1910e8c4075aad53f210f7e9a85 : SUCCESS in 8m 38s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/65ff804e7f9c4725a6b7cccea09f5f16 : FAILURE in 10m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ec276cd77264e16ad6b646d4f24a174 : SUCCESS in 37m 48s","accounts_in_message":[],"_revision_number":1},{"id":"224f9e892b71dc6bc9ee5a270ca060214d52841f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-07 07:20:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24333d5e62b85f7cb781352f7e0ec41aef0e576d","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-03-08 15:27:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"da93cccefc0a444390f5cf63e8d18f847c142089","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-03-12 14:42:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d71473a49e75281fe9e31b859b5c2c54d11f7f6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 14:44:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5981e4c891d205e633b341308b129a0a6c791b2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 17:22:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9981f08e46224b849f707efeb0f5a469\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01b0794c8a1f4aa1bccb4a79c2aa9fe1 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a91e316634a84b83b7f3ecf9ad0a4cc3 : SUCCESS in 10m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac78e2adc3ee461f8173e2c42ba8e0ae : SUCCESS in 10m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23bb61be16574ff58f905d6025f30819 : SUCCESS in 12m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/40eb684c4cb848fb95cafa4efdea9833 : SUCCESS in 1h 03m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/01e64145103041a883c144b62c7d1575 : SUCCESS in 1h 13m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cad89aa34c241b3bc1831f599761c08 : SUCCESS in 9m 43s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/04d098094f91403b992a069827367966 : SUCCESS in 16m 48s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/c00f1a4ce5ff47c3a8eea13f5c301929 : SUCCESS in 22m 46s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/1b8b524c44c94ff38ee41bbda439ebe4 : SUCCESS in 34m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b939ca77c6c64e42ac296f820516455f : SUCCESS in 1h 02m 27s","accounts_in_message":[],"_revision_number":1},{"id":"6006dc81c6cfe66cf588e04e9cb78a9c19487774","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 17:22:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d4786f1a0fc0e39837fa9a7afd200ea4177e7523","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 17:26:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4de6cacdb8a4fdda91e807d7f76c6fe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8efb11f6f2284eed8b635b2a4d2c3dfc : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39bb2ee646294c8bbb9d432ad0e212aa : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1},{"id":"ff218f25315f20189bc6b27148cf5cafb99f1656","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-13 13:04:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c7020c51ab2c42ec65f3e916c03f40ebeac2c4e","revisions":{"6c7020c51ab2c42ec65f3e916c03f40ebeac2c4e":{"kind":"REWORK","_number":1,"created":"2024-01-27 12:03:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/906953/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/53/906953/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/53/906953/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/53/906953/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/53/906953/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/53/906953/1"}}},"commit":{"parents":[{"commit":"0608537f034b4b71205a90c7fbf02aa8b6720c8c","subject":"Merge \"Check user existence before setting last_active_at\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0608537f034b4b71205a90c7fbf02aa8b6720c8c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-27 11:53:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-27 12:02:14.000000000","tz":540},"subject":"Drop remaining references to eventlet options","message":"Drop remaining references to eventlet options\n\nBecause these were removed by [1]. Also update the previous release\nnote to document the upgrade impact on catalog information (like\nendpoint urls) including string interpolations requiring these removed\noptions.\n\n[1] 2a3c73c49b117fe43d2174dbdb55842a4407377d\n\nChange-Id: If78d0b93665410b86754ea35653ca9d4c15c81c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6c7020c51ab2c42ec65f3e916c03f40ebeac2c4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6c7020c51ab2c42ec65f3e916c03f40ebeac2c4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
