)]}'
{"id":"openstack%2Fkeystone~907755","triplet_id":"openstack%2Fkeystone~master~Ida6e6fbe61358dca3e339b69b321d6dbcce28382","project":"openstack/keystone","branch":"master","topic":"bug/NotificationsTestCase-isolation","attention_set":{"935":{"account":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"last_update":"2024-02-05 15:38:11.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ida6e6fbe61358dca3e339b69b321d6dbcce28382","subject":"tests: Ensure oslo_messaging_notifications group exists","status":"NEW","created":"2024-02-05 15:17:06.000000000","updated":"2024-12-13 21:29:55.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9475e9c2c6d807c32a478c2ade26232a06b029b4","_number":907755,"virtual_id_number":907755,"owner":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-05 17:14:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-05 15:29:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-05 15:38:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-13 21:29:55.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"}],"messages":[{"id":"6c05becb8753e5f83759f1d643947e7cc824cd11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-02-05 15:17:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1868cfbf194a12a412113d20aac02b3e32ebb039","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 15:29:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8955058ec2eb45b39f0381fc41d3fa2f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1fedcd640da4e5081cddadfda712067 : SUCCESS in 10m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f6bc7eb1879435e96a05f442219444c : SUCCESS in 11m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb07d56ed3f18b6e8f558b8e6702d9ef8def6974","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 15:38:11.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/94c19dfd278c46bf91f15a4db6bc8d37\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/165bf5f30d5540a193de2977f4c0e29b : SUCCESS in 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cf87fcc6ce6492f95443558306a8ead : FAILURE in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e585bdd437e34f238f2ebc7448eebc3a : SUCCESS in 13m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d90f0ace27b4a148d6ef5f620e3e7fb : SUCCESS in 12m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03120dd04e6645639efbd92737af7f5a : SUCCESS in 11m 31s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f523745c37684457af1b524fd24a9960 : SUCCESS in 14m 45s","accounts_in_message":[],"_revision_number":1},{"id":"74c6e359ff7b20aa6af41bca632a8227a21c9427","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-02-05 16:48:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7ce01af215f112049189e6c941ee4e954e27135b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 17:14:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d98b2f98bbb9466bbb84aa82ad15c18a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36f15540d3684adaa999b15dff016a14 : SUCCESS in 14m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0ca69a22eb747d7a253b6c576137e71 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7be976a2a15242b4befafc708e9d7ea6 : SUCCESS in 9m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/12a27d949caa4ceca99168ee74139084 : SUCCESS in 11m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a698111a9454b7db314e8f3ac611d99 : SUCCESS in 13m 44s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ee39e2e70c4440f8834c242b7ccde8cb : SUCCESS in 11m 56s","accounts_in_message":[],"_revision_number":2},{"id":"a7d243f7b83238c94fab531e59c63fdd20977932","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 17:33:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4904b131c09c4da780d3316ca04eb056\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0fcc7d67ef124e299992588c03672d20 : SUCCESS in 40m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6dbd2cbb767f4fe1961de5e39165e8c5 : SUCCESS in 11m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9475e9c2c6d807c32a478c2ade26232a06b029b4","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-12-13 21:29:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"33c60b4ca31588cbdf91af4eb07be2c2264bec5e","revisions":{"72c600bd90baf7f2dfa1cfb7b523ce07ab33d313":{"kind":"REWORK","_number":1,"created":"2024-02-05 15:17:06.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/55/907755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/55/907755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/55/907755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/55/907755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/55/907755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/55/907755/1"}}},"commit":{"parents":[{"commit":"03401210426ed62bbf24c24793125b598321f182","subject":"Allow users with \"admin\" role to get projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/03401210426ed62bbf24c24793125b598321f182"}]}],"author":{"name":"James Page","email":"james.page@canonical.com","date":"2024-02-05 15:16:09.000000000","tz":0},"committer":{"name":"James Page","email":"james.page@canonical.com","date":"2024-02-05 15:16:09.000000000","tz":0},"subject":"tests: Ensure oslo_messaging_notifications group exists","message":"tests: Ensure oslo_messaging_notifications group exists\n\nCONF is a global object and dependening what order unit tests are\nexecuted and in which stestr worker process the oslo_messaging_notifications\nconfiguration group may not actually be present at the point in\ntime the configuration fixture is created.\n\nEnsure that this group is present by setting the notification transport\nduring test setup.\n\nChange-Id: Ida6e6fbe61358dca3e339b69b321d6dbcce28382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/72c600bd90baf7f2dfa1cfb7b523ce07ab33d313"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/72c600bd90baf7f2dfa1cfb7b523ce07ab33d313"}]},"branch":"refs/heads/master"},"33c60b4ca31588cbdf91af4eb07be2c2264bec5e":{"kind":"REWORK","_number":2,"created":"2024-02-05 16:48:20.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/55/907755/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/55/907755/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/55/907755/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/55/907755/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/55/907755/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/55/907755/2"}}},"commit":{"parents":[{"commit":"03401210426ed62bbf24c24793125b598321f182","subject":"Allow users with \"admin\" role to get projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/03401210426ed62bbf24c24793125b598321f182"}]}],"author":{"name":"James Page","email":"james.page@canonical.com","date":"2024-02-05 15:16:09.000000000","tz":0},"committer":{"name":"James Page","email":"james.page@canonical.com","date":"2024-02-05 16:48:12.000000000","tz":0},"subject":"tests: Ensure oslo_messaging_notifications group exists","message":"tests: Ensure oslo_messaging_notifications group exists\n\nCONF is a global object and dependening what order unit tests are\nexecuted and in which stestr worker process the oslo_messaging_notifications\nconfiguration group may not actually be present at the point in\ntime the configuration fixture is created.\n\nEnsure that this group is present by setting the notification transport\nduring test setup.\n\nChange-Id: Ida6e6fbe61358dca3e339b69b321d6dbcce28382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/33c60b4ca31588cbdf91af4eb07be2c2264bec5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/33c60b4ca31588cbdf91af4eb07be2c2264bec5e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
