)]}'
{"id":"openstack%2Fkeystone~908163","triplet_id":"openstack%2Fkeystone~master~Ib537f1dd33dd3f3dc8909873dffc37980d04b4db","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2024-03-12 14:39:46.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2024-03-12 17:07:39.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-03-12 17:07:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib537f1dd33dd3f3dc8909873dffc37980d04b4db","subject":"Fix federation mapping role jsonschema","status":"MERGED","created":"2024-02-06 16:35:47.000000000","updated":"2024-03-12 17:10:03.000000000","submitted":"2024-03-12 17:07:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908163","meta_rev_id":"a7ec424bdcc932c4267079124f29763e4fc736b5","_number":908163,"virtual_id_number":908163,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-12 17:07:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2024-03-08 15:50:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2024-03-12 14:39:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2024-03-12 14:39:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-06 17:23:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-06 18:31:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-08 15:50:19.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2024-03-08 15:50:19.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-03-08 15:50:19.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"4332c35f8e10756445a5511ea0b1d76c6c587d76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-02-06 16:35:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"faa8c8cf5c7e909994f9ca2c13ac8731eea7dadc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 17:23:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3bf311c3f464787b4e02be215ebef98\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/45f504bb9beb4599a3a2d4bed3624e9f : SUCCESS in 10m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f5cfec3b8a84e9381f79a50ee89cb54 : SUCCESS in 42m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3bceab7e7a59c5ca5352e0f367694df926b696a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 18:31:18.000000000","message":"Patch Set 1: Verified-1\n\n(23 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/020e2b756fb14167b6cf1f06e1cac2e5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27bee542ac8e46f98872d861eb81d743 : SUCCESS in 14m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36a4f60e8fbf41a4ab4666fc6a5d9182 : FAILURE in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5704bae3b4314eba89981a27ff3a8d15 : SUCCESS in 11m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/789c55e449b844d0844833dc0e9ba87b : SUCCESS in 12m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a406bbf7845f41d4a26e75e0fc087fac : SUCCESS in 11m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/2e36693973724b7ca6ffe62204600dbc : SUCCESS in 1h 17m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2aa85905c8e4e078fa85ff61d9b785f : SUCCESS in 1h 46m 33s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1f16a314a1864b7d8a6563dee7b3f4ee : SUCCESS in 14m 03s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/716c024a20434385936693ba68d27e21 : SUCCESS in 39m 36s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/039865e566114c3ebdfc4c8c56415171 : SUCCESS in 48m 21s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4c67f21e451428893309b2eaf3467c1 : FAILURE in 39m 19s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/cc9edeacb046410cb342fccfbd8c0921 : SUCCESS in 26m 36s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/30c97683f6a744adb73b2e3baf5e47b1 : SUCCESS in 11m 32s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/f8d8b60b046e4317ab2541c6dab41418 : FAILURE in 27m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb6b96ad41664f789b7237fbaf2da325 : SUCCESS in 1h 05m 20s","accounts_in_message":[],"_revision_number":1},{"id":"6bb3e7849431dcd531af2a46acb83d33893773aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-02-07 07:50:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f3ba19f804158119f54024d373c429f0df510fce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 08:37:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d035febe1904bd3ac76b4e0312c69ed\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/959d221326474ffaba621016f5738fd4 : SUCCESS in 45m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/851c7d9217dc4bdba57927c3b79f80a9 : SUCCESS in 41m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7b094707c291e39023321a9ae8e973dfaefba64a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 09:27:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/845db9b2d852422789c065312d9d5a08\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70b51292068447bba620b89aeeba8158 : SUCCESS in 12m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/972c7dd0aaf844a79ea3ef93b6eab13e : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cef07aee85324da38d82ce327c9f1673 : SUCCESS in 11m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd6a2d1810c744849119749626d457b9 : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/487aefcd67e348fc9934b38d843ee0de : SUCCESS in 12m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/e3a68f6b51764c90a823d3ee679170c3 : SUCCESS in 1h 04m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7ff13e76fdfe434b8aac9ee83db153fd : SUCCESS in 1h 34m 41s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ef2860af4eb04400a3362be9e64b69c8 : SUCCESS in 13m 19s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/564e1f1d696342a594ba1da40956e8e0 : SUCCESS in 25m 31s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/c48d7cfadf864947afe8bc31f7f17661 : SUCCESS in 46m 39s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b080fdda1644c50a24ed4d8b95a8b07 : FAILURE in 26m 17s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/c1967c4a7d6749ad9f5bdb9c80d99b9c : SUCCESS in 36m 34s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/dd8999f2096f4b92a12f27769507fda6 : SUCCESS in 9m 29s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/518c1ec64b874ed7aa7c1c5536de922c : FAILURE in 24m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55c64ca22cf74890a0dd4bf34a2423f1 : SUCCESS in 40m 55s","accounts_in_message":[],"_revision_number":2},{"id":"1208bdaac74065e760251ca672f5196834b61721","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-03-08 15:50:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d95b9763c6160a889a461a46d8e82898f01317e1","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-03-12 14:39:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ed0a783bb08722d71dd28e97adbf2573f0f12d2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 14:40:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cb316511f23b1a51e1de85c0c67d89389ae8879f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 17:07:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f4ffc8ffc3848c0a1428cda8018d4fb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc9d3921357f476ba08866c95b0a04cb : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bb3adb7f59b4ba09f515fed3779a8e7 : SUCCESS in 11m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dcf0f85ec42140398fa6b4b39576dbbe : SUCCESS in 11m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f7e08a44bb247a6ae55d5967cfe05cb : SUCCESS in 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/029faf6097b744aab77ad4e77e58b803 : SUCCESS in 52m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3772808a61eb4ae3a6079dd2e9945c1c : SUCCESS in 1h 24m 27s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/15dd114b8aa84138be39dfe5f5d7e15a : SUCCESS in 16m 45s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/64380fa2eb2443d3bcbc70c81e941ef7 : SUCCESS in 38m 35s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/9e2c6cddc14a494392d171adb39732d4 : SUCCESS in 30m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/beb17f4237fb405695ac80d72cf0e5f3 : SUCCESS in 42m 32s","accounts_in_message":[],"_revision_number":2},{"id":"a6a9dbe4ded3897d54a64b748da69ac1f9eb7367","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 17:07:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a7ec424bdcc932c4267079124f29763e4fc736b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 17:10:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ddc0a1e88624c5b92df27743d67b2f4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaf3430887e7474a8160e4a04b83fbc7 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"480ba297095490b43d89f4fe93603b49e0d6f430","revisions":{"4b3a2a89975ee568f9488f5f446be516fbd16c63":{"kind":"REWORK","_number":1,"created":"2024-02-06 16:35:47.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/63/908163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/63/908163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/63/908163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/63/908163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/63/908163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/63/908163/1"}}},"commit":{"parents":[{"commit":"03401210426ed62bbf24c24793125b598321f182","subject":"Allow users with \"admin\" role to get projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/03401210426ed62bbf24c24793125b598321f182"}]}],"author":{"name":"gtema","email":"artem.goncharov@gmail.com","date":"2024-02-06 16:31:00.000000000","tz":60},"committer":{"name":"gtema","email":"artem.goncharov@gmail.com","date":"2024-02-06 16:35:44.000000000","tz":60},"subject":"Fix federation mapping role jsonschema","message":"Fix federation mapping role jsonschema\n\nadditionalProperties attribute must be located on the level of \"type\"\nand not inside \"properties\"\n(https://json-schema.org/understanding-json-schema/reference/object#additional-properties).\nSadly this is not violating schema validation, but is wrong and hurts\nany reasonable processing of the schema.\n\nChange-Id: Ib537f1dd33dd3f3dc8909873dffc37980d04b4db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4b3a2a89975ee568f9488f5f446be516fbd16c63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4b3a2a89975ee568f9488f5f446be516fbd16c63"}]},"branch":"refs/heads/master"},"480ba297095490b43d89f4fe93603b49e0d6f430":{"kind":"REWORK","_number":2,"created":"2024-02-07 07:50:03.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/63/908163/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/63/908163/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/63/908163/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/63/908163/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/63/908163/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/63/908163/2"}}},"commit":{"parents":[{"commit":"03401210426ed62bbf24c24793125b598321f182","subject":"Allow users with \"admin\" role to get projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/03401210426ed62bbf24c24793125b598321f182"}]}],"author":{"name":"gtema","email":"artem.goncharov@gmail.com","date":"2024-02-07 07:49:45.000000000","tz":60},"committer":{"name":"gtema","email":"artem.goncharov@gmail.com","date":"2024-02-07 07:49:45.000000000","tz":60},"subject":"Fix federation mapping role jsonschema","message":"Fix federation mapping role jsonschema\n\nadditionalProperties attribute must be located on the level of \"type\"\nand not inside \"properties\"\n(https://json-schema.org/understanding-json-schema/reference/object#additional-properties).\nSadly this is not violating schema validation, but is wrong and hurts\nany reasonable processing of the schema.\n\nChange-Id: Ib537f1dd33dd3f3dc8909873dffc37980d04b4db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/480ba297095490b43d89f4fe93603b49e0d6f430"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/480ba297095490b43d89f4fe93603b49e0d6f430"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
