)]}'
{"id":"openstack%2Fkeystone~908195","triplet_id":"openstack%2Fkeystone~master~Iae525ee13dec72af6a7d70db2bb59a77c682a177","project":"openstack/keystone","branch":"master","topic":"user-delete-initiator","attention_set":{},"removed_from_attention_set":{"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2024-03-12 17:07:37.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2024-03-12 14:39:10.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2024-03-12 17:07:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iae525ee13dec72af6a7d70db2bb59a77c682a177","subject":"Pass initiator to delete user","status":"MERGED","created":"2024-02-06 19:59:18.000000000","updated":"2024-03-12 17:08:58.000000000","submitted":"2024-03-12 17:07:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908195-user-delete-initiator","meta_rev_id":"8ca369adb47493a909590fe2f5a3b665800d1876","_number":908195,"virtual_id_number":908195,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-12 17:07:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2024-03-08 15:52:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2024-03-12 14:39:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2024-03-12 14:39:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-06 20:11:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-06 21:50:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-08 15:52:55.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2024-03-08 15:52:55.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-03-08 15:52:55.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"d7de8cff10861b0352ef9924a9447801c07cfac6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-02-06 19:59:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c540d9ab1918721df7fbd95130bac73acbd531b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 20:11:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b0cd6fec99e4bcea6e6c738a9f35e8b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e73a6671e1b64a579f06564dd0b2899e : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/08ff117fc812485583cd09fc06feb703 : SUCCESS in 11m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f78b69f7b03d5d926d01e5588adf47715768b10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 21:50:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cda578eeec04afa8a04946b541eef62\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b0327cddb5c47a594833f06d6f87eef : SUCCESS in 16m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8fefb73f04e49c4a26dbd1ab61e1540 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b74042ef5a394a6488b44cecfc2e6082 : SUCCESS in 11m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/493d8cf2d3754f1bad6beb805e2ce69e : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e264c86b38374d4b8f9b0737e684fdd5 : SUCCESS in 10m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/dc28368752ef4689aca095a0391d512d : SUCCESS in 55m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eda183fb35174890a25b0b77ebfb289a : SUCCESS in 1h 50m 32s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b1bc30194883453ab5cc648b7960fa53 : SUCCESS in 14m 07s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f4232eb8f83d4baca070f883c1f3573e : SUCCESS in 22m 10s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/729d7dbe362347e0970101ccf66a8d55 : SUCCESS in 47m 19s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3726e1ab9dcc4dfebc2db1e3038f8fbe : FAILURE in 24m 06s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/797bdba7fbd440f7b22d471efd447af2 : SUCCESS in 33m 25s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/483435af1549481ba672bc6dca958490 : SUCCESS in 11m 03s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/4a86a30b5c234c82aa7b936373d00613 : FAILURE in 12m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7931583d1cf340c7979251e597852cab : SUCCESS in 1h 01m 25s","accounts_in_message":[],"_revision_number":1},{"id":"9010c442dbccf707490dbe8955f93614dfab23fc","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-03-08 15:52:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c7d031565e36af4d3560a733528bc607a81c2481","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-03-12 14:39:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f550587871645c77b0c180cf8b1ad0fee2462f8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 14:39:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c65d4b007bc22c00c6820eb81dfc3857118776f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 17:07:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18e5e88e284947eaa856fa63bb3d6c28\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a1ef9e796964a6cb85f9909e7650677 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c67ede7de302446b8f3391466c988e28 : SUCCESS in 11m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f4102a739c614de9b6702f56f83dd6df : SUCCESS in 10m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e6b60fede784aee86fee3c7ebea3501 : SUCCESS in 11m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/df857ef9a1764b5da32718cfe84b0b57 : SUCCESS in 52m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a035e4d45a44ed1b6e5543e90fd9030 : SUCCESS in 1h 28m 29s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b26c2742827444c298374ea7464e8641 : SUCCESS in 13m 26s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d6f3a8a2bddc40a48419180a259dd427 : SUCCESS in 34m 48s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/4e0712c7159d45798a15810873d27e2b : SUCCESS in 34m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd5a4674020c4bf08ba74c02ab126171 : SUCCESS in 51m 19s","accounts_in_message":[],"_revision_number":1},{"id":"1515bcf8fee2ff909956af47717d3be489999de6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 17:07:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8ca369adb47493a909590fe2f5a3b665800d1876","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 17:08:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a1c022f8cbb4e9baee166a357bd18b6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b63b2aaf8fe54657853422d9835ffdf9 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2735a28fc5a12bba5252383dfb5e68a0b99f41f9","revisions":{"2735a28fc5a12bba5252383dfb5e68a0b99f41f9":{"kind":"REWORK","_number":1,"created":"2024-02-06 19:59:18.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/95/908195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/95/908195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/95/908195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/95/908195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/95/908195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/95/908195/1"}}},"commit":{"parents":[{"commit":"03401210426ed62bbf24c24793125b598321f182","subject":"Allow users with \"admin\" role to get projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/03401210426ed62bbf24c24793125b598321f182"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-02-06 19:50:26.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-02-06 19:56:39.000000000","tz":0},"subject":"Pass initiator to delete user","message":"Pass initiator to delete user\n\notherwise the initiator field is missing from the CADF payload,\nwhich misses the point of audit and technically makes these notifications\nnot valid as CADF events (initiator field is requires by the\nCADF spec).\n\nChange-Id: Iae525ee13dec72af6a7d70db2bb59a77c682a177\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2735a28fc5a12bba5252383dfb5e68a0b99f41f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2735a28fc5a12bba5252383dfb5e68a0b99f41f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
