)]}'
{"id":"openstack%2Fkeystone~908763","triplet_id":"openstack%2Fkeystone~master~Ibbde313ac8bcf3187c139bbd0840702f229534d0","project":"openstack/keystone","branch":"master","topic":"ec2-fixes","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-06-28 17:14:41.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2024-03-12 14:41:57.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-06-28 17:14:41.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2024-06-28 15:22:13.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}},"hashtags":[],"change_id":"Ibbde313ac8bcf3187c139bbd0840702f229534d0","subject":"Add test with noauth for s3tokens and ec2tokens","status":"MERGED","created":"2024-02-12 09:28:32.000000000","updated":"2024-06-28 17:18:00.000000000","submitted":"2024-06-28 17:14:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908763-ec2-fixes","meta_rev_id":"eb3a087d5c6b5bb29316ecfa988213b613b77ccb","_number":908763,"virtual_id_number":908763,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-28 17:14:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2024-03-18 12:42:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2024-03-29 07:36:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2024-06-28 15:22:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2024-06-28 15:22:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-12 09:46:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-12 09:50:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-08 15:39:12.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2024-03-08 15:39:12.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-03-08 15:39:12.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-03-29 07:36:54.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"f8fe0b75773fae96a45ec52a1329df87d916aa12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-02-12 09:28:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"894062d9020b494452a08759598bd5cd63c219a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 09:46:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/403957dea6f74f19984c9a11cf35a311\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3222abf1659b4d0a9861be0435e17171 : SUCCESS in 9m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/16d4314477654a0cb1050f539a20f5b0 : SUCCESS in 17m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e7aee68457061d0f8268f1d1e24d8dfec647b02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 09:50:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/861df0eff5a348dc8bc4c69213f7b956\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56ad9298951747318a9b23787f14d520 : SUCCESS in 16m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7f1a9e3fed34d89a3b4224fccf55ab8 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd81283317a3440185f4e8689564d265 : SUCCESS in 9m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d4f203117a94df497179e902c77eeb5 : SUCCESS in 9m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec6790eebfaf4038a49cdc67320e08f8 : SUCCESS in 11m 46s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/38ef357633ae438a8fa8da0771c00105 : SUCCESS in 13m 45s","accounts_in_message":[],"_revision_number":1},{"id":"3f86bccd316cb92ae9a127341f87944eb0c759f5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-02-19 15:17:43.000000000","message":"Topic set to ec2-fixes","accounts_in_message":[],"_revision_number":1},{"id":"7255cf47d86d6ecb6378a79a04b5f423b3b3ef6e","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-03-08 15:39:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"adfcfd6196083f44505068b7b5664145d120c137","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-03-12 14:41:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"77436a582cefbadf2e8297c7702946c95b8548ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-03-18 12:38:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"deec8aa360b0dc03a729e0f2a0bd2c08192483ec","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-03-18 12:42:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2802c90519ea824fb998f12509fb301ada44e24f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 12:58:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff86c3fb6e964523af0c5fe54be6e673\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49b844cf5f7c4f1cabcb9ba01db2d423 : SUCCESS in 16m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b0588588a74a43788efaece44599c227 : SUCCESS in 16m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3d038ae12565450fb8a474c3d8ee05e4 : SUCCESS in 13m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a90c0d93132e6e0ad42753cbdb215ea4270ce048","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 13:10:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efa4cebf411644b693e11d209575fae7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca2150f7b389459c9ebc8fe0bc411b27 : SUCCESS in 11m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50fc4d8575464d2a8a6da5c714263c51 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f434f33c6a8401ab5a75a7a8174618e : SUCCESS in 10m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd31ae5669e54c4a9ae7668dd3dc9f6e : SUCCESS in 11m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e750821e03bb49b49eb45f21c931a090 : SUCCESS in 12m 53s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e30ec9ab66c34b1daaa85851796d5153 : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":2},{"id":"b3adcb416deaa245c37b522dbba85fdeae6d3f9a","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2024-03-29 07:36:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1850b8d9d4c48b44b31282cb326b0ceb758ace06","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-06-28 15:22:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9277dfbe8ed1c89a76589a13ff1f3177e0c8aabd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 15:22:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6478d599af49e48150c9aef29b6eaaa3c67ef010","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 17:14:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c17d1ea33cd34626b61f78537a8e2941\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17736af6aa9e4c28812e75fa837d04d9 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e426f00a92244199032b39683a5e323 : SUCCESS in 12m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0700874834a459abe9bde81f518831e : SUCCESS in 10m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee0466df39e248cfa4c5de8bb79f2c36 : SUCCESS in 10m 31s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/888059de21ec420da85ff4ebf47e081a : SUCCESS in 37m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e63702af4ddf3579e8c0cce14a8bdc5e3f4f89ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 17:14:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"eb3a087d5c6b5bb29316ecfa988213b613b77ccb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 17:18:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5e566917ef04c5294b425a0406fb219\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1599eaed373d4866976cdb9179a47e28 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bd70653a2445fe1b9628ff1f92cbc58f2c9f4d70","revisions":{"b2a996d2db9a0f3bff598033d2dce9a28ae16ffb":{"kind":"REWORK","_number":1,"created":"2024-02-12 09:28:32.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/63/908763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/63/908763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/63/908763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/63/908763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/63/908763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/63/908763/1"}}},"commit":{"parents":[{"commit":"db0ff104763b6da4d661bf0c5cc9814ea3f18fc8","subject":"reno: Update master for unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/db0ff104763b6da4d661bf0c5cc9814ea3f18fc8"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2024-02-12 09:26:35.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2024-02-12 09:26:35.000000000","tz":0},"subject":"Add test with noauth for s3tokens and ec2tokens","message":"Add test with noauth for s3tokens and ec2tokens\n\nThe s3tokens and ec2tokens endpoint is a public API\nthat does not require authentication to be called but\nthat is what the testing does today.\n\nThis duplicates the \"good request\" tests by also\ntesting with noauth set so that we don\u0027t send a\ntoken in the request, this can prevent bugs where\nwe would unintentionally require auth for these\nendpoints.\n\nChange-Id: Ibbde313ac8bcf3187c139bbd0840702f229534d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b2a996d2db9a0f3bff598033d2dce9a28ae16ffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b2a996d2db9a0f3bff598033d2dce9a28ae16ffb"}]},"branch":"refs/heads/master"},"bd70653a2445fe1b9628ff1f92cbc58f2c9f4d70":{"kind":"REWORK","_number":2,"created":"2024-03-18 12:38:44.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/63/908763/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/63/908763/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/63/908763/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/63/908763/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/63/908763/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/63/908763/2"}}},"commit":{"parents":[{"commit":"4121cf6cb7c3d5e585df827d72e08296664f7326","subject":"Merge \"Add ability to create users and projects from keystone-manage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4121cf6cb7c3d5e585df827d72e08296664f7326"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2024-02-12 09:26:35.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-03-18 12:38:42.000000000","tz":60},"subject":"Add test with noauth for s3tokens and ec2tokens","message":"Add test with noauth for s3tokens and ec2tokens\n\nThe s3tokens and ec2tokens endpoint is a public API\nthat does not require authentication to be called but\nthat is what the testing does today.\n\nThis duplicates the \"good request\" tests by also\ntesting with noauth set so that we don\u0027t send a\ntoken in the request, this can prevent bugs where\nwe would unintentionally require auth for these\nendpoints.\n\nChange-Id: Ibbde313ac8bcf3187c139bbd0840702f229534d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bd70653a2445fe1b9628ff1f92cbc58f2c9f4d70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bd70653a2445fe1b9628ff1f92cbc58f2c9f4d70"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
