)]}'
{"id":"openstack%2Fkeystone~908850","triplet_id":"openstack%2Fkeystone~master~I0a87cc5a35e6033d670bab56d5cdc8b8312819d8","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-03-12 16:59:59.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2024-03-12 14:38:14.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-12 16:59:59.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2024-03-12 16:59:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0a87cc5a35e6033d670bab56d5cdc8b8312819d8","subject":"Remove unused old job templates and experimental jobs","status":"MERGED","created":"2024-02-12 17:46:08.000000000","updated":"2024-03-12 17:02:33.000000000","submitted":"2024-03-12 16:59:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908850","meta_rev_id":"dc466b2b56c0fcff5d991f9507dcbd339cde362b","_number":908850,"virtual_id_number":908850,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-12 16:59:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2024-03-12 14:38:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2024-03-12 14:38:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-12 18:32:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-12 19:07:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-07 07:20:16.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2024-03-07 07:20:16.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-03-07 07:20:16.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"ffacbdec74e71178d74c0acfb0891e92196cdfeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-12 17:46:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e652a5b8c2ccdce80112c6547dabbf8faeaf463f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-12 17:47:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5b66572854c2a590817f757709706908b4ff2e09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 18:32:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17609b29eb104cf592bddc18fbfbb2c0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1662f7b8b2b4590a57a3f83854cbc0b : SUCCESS in 42m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b7e4f442665e4b1abd34bc02252f85b6 : SUCCESS in 37m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"46476e9ca1a20eed88c7ab10632f24bd966e204a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 19:07:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4a5acc3dc204da2834ad1700bad7553\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efc69402ebc44da794a6dbf0768f51af : SUCCESS in 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85c1f8fefdd54e26923fca420f2c2847 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b114c469d2343318d225851665fe163 : SUCCESS in 11m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7945ca8041a44082b82d37d546bbd299 : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4242d6dead74fcba91ecd9095d9a409 : SUCCESS in 12m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/ada64d9233534ae7be8c94fbf185dc45 : SUCCESS in 1h 05m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e75d8bff824347fda4b21af963023e84 : SUCCESS in 1h 15m 38s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3d6de16341144c73b4215a1ae9e10e86 : SUCCESS in 13m 32s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/038df87d56304e62b99f0b722c5d820b : SUCCESS in 36m 44s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/004b4b3c22444e91b911af0fa6dbad5c : SUCCESS in 30m 53s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/628ca5d03b5e4ab9bc777407a573effd : FAILURE in 36m 29s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/4836fe349b4a450f843ff8212f83282f : SUCCESS in 22m 42s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/5b94dd351fd148bcb9aca7d07f4e8e2c : SUCCESS in 10m 12s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ffdd2abae251455f9bfadeb342a1e5b3 : FAILURE in 26m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d8eb45badba42b699cc03d06741e99e : SUCCESS in 44m 56s","accounts_in_message":[],"_revision_number":2},{"id":"12e9d98a9107892ed0ee855d026b1b0cc080ec3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-03-08 15:30:01.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":3},{"id":"162efa9ddc69b751296cc48f4bcfc6acbe41ba82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 15:43:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3a8e58928bc448493f32186b6a750ba\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c28b26c9f954c308f60e0a7a071b533 : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8450518f0f8649cabeb3d3c1ab348157 : SUCCESS in 12m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9605c6ae732cc6bd87a5850c115aa9a049efb61a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 17:24:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45e869950a334c3191c46677f78ccb14\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ee9b61a143c4b9aaf25eed9e6a9d6d8 : SUCCESS in 14m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/369bb363592c4fb3b11acd94585a2652 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00a397a8319946f0b1879d5224cf1a00 : SUCCESS in 10m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5508e914920443e4ad7df4728667a3a9 : SUCCESS in 11m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00ae7c04ac71472ab9f898ace0717b7d : SUCCESS in 11m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/3faec86288dd48789dd2cfb71d39a40e : SUCCESS in 1h 04m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a6e0f42a4a1c4abcb421652fd53e595e : SUCCESS in 1h 47m 30s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/098a61deb1c244bbb35eebd3a7d05a92 : SUCCESS in 11m 59s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8066cb7715d44a9f8e88f2e03bf85f36 : SUCCESS in 45m 15s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/59fe252cfed748f193a07fc680c49e60 : SUCCESS in 29m 31s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87a6dc1393ab462c93e1fb19218b4658 : FAILURE in 27m 18s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/9f959df701c942a3ae5637481be53900 : SUCCESS in 24m 09s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/4e865e8b53e44bc78507d6aedd021fa9 : SUCCESS in 26m 07s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/22ae8e38b9824bd3a2b91ef44fe66287 : FAILURE in 13m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5eb20806b2304da388a9c3dc8c28f186 : SUCCESS in 1h 04m 52s","accounts_in_message":[],"_revision_number":3},{"id":"a0596d4c2f71d26972a94561d82aa220ba7c8121","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-03-12 14:38:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c5725db5365d931fc36c8574e6926b69eeb28851","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 14:39:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"499106f1ebdaa7de6b4e1bab62d86d3cc3d40dcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 16:59:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2690b7992a9747f9a9717f42b9e41ce9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e132404d48224c6f938b8bfade589142 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a09287d8516640c08deb3ac732aeefc6 : SUCCESS in 11m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c15f8ace12f8490686ffa644a6e8d714 : SUCCESS in 11m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8c2c6ec766d4864b0af3779d98caa31 : SUCCESS in 10m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/42982ca7ff604f379544ed95bc34b0f5 : SUCCESS in 1h 16m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c508c77757fc4878a2a400d14e5c06ec : SUCCESS in 1h 16m 54s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e4fca1fa9ab343289592c7ad35404b7c : SUCCESS in 11m 34s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1882ae0505bb4debbae5100d65e84ad9 : SUCCESS in 32m 55s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/720d1d72e7a54909acadc5c9e0e2dbb0 : SUCCESS in 25m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f37c0075f074bf6a71469eb32fc5671 : SUCCESS in 50m 20s","accounts_in_message":[],"_revision_number":3},{"id":"8faedc026224f56ec9e2e8a7fa5e7cdef9cdf332","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 16:59:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"dc466b2b56c0fcff5d991f9507dcbd339cde362b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 17:02:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cc65f0e7ad6432ab9325070bb3bc0a6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5337fbed09894acf9ec8c29951d48285 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"609ec29ac95817941543646d05bf783a3774ea77","revisions":{"450f06acdc75823b56a263e206b045814151ff52":{"kind":"REWORK","_number":1,"created":"2024-02-12 17:46:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/908850/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/50/908850/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/50/908850/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/50/908850/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/50/908850/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/50/908850/1"}}},"commit":{"parents":[{"commit":"5a97b7d847d5471d91b7e41ab0acf65974419c44","subject":"Merge \"Fix policies for groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5a97b7d847d5471d91b7e41ab0acf65974419c44"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-12 17:45:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-12 17:45:51.000000000","tz":540},"subject":"Remove unused old job templates and experimental jobs","message":"Remove unused old job templates and experimental jobs\n\nChange-Id: I0a87cc5a35e6033d670bab56d5cdc8b8312819d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/450f06acdc75823b56a263e206b045814151ff52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/450f06acdc75823b56a263e206b045814151ff52"}]},"branch":"refs/heads/master"},"31dd695053eabdae15f1fdd7752ea664484de7b7":{"kind":"REWORK","_number":2,"created":"2024-02-12 17:47:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/908850/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/50/908850/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/50/908850/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/50/908850/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/50/908850/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/50/908850/2"}}},"commit":{"parents":[{"commit":"5a97b7d847d5471d91b7e41ab0acf65974419c44","subject":"Merge \"Fix policies for groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5a97b7d847d5471d91b7e41ab0acf65974419c44"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-12 17:45:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-12 17:47:18.000000000","tz":540},"subject":"Remove unused old job templates and experimental jobs","message":"Remove unused old job templates and experimental jobs\n\nUbuntu Xenial, CentOS7 and OpenSUSE 15 are all too old.\n\nChange-Id: I0a87cc5a35e6033d670bab56d5cdc8b8312819d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/31dd695053eabdae15f1fdd7752ea664484de7b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/31dd695053eabdae15f1fdd7752ea664484de7b7"}]},"branch":"refs/heads/master"},"609ec29ac95817941543646d05bf783a3774ea77":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-03-08 15:30:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/50/908850/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/50/908850/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/50/908850/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/50/908850/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/50/908850/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/50/908850/3"}}},"commit":{"parents":[{"commit":"da1c884b90c2f45c131bc45cd06e90e5f83e70fe","subject":"Drop keystone-dsvm-functional-federation-opensuse15 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/da1c884b90c2f45c131bc45cd06e90e5f83e70fe"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-12 17:45:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-08 15:30:01.000000000","tz":0},"subject":"Remove unused old job templates and experimental jobs","message":"Remove unused old job templates and experimental jobs\n\nUbuntu Xenial, CentOS7 and OpenSUSE 15 are all too old.\n\nChange-Id: I0a87cc5a35e6033d670bab56d5cdc8b8312819d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/609ec29ac95817941543646d05bf783a3774ea77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/609ec29ac95817941543646d05bf783a3774ea77"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
