)]}'
{"id":"openstack%2Fkeystone~909238","triplet_id":"openstack%2Fkeystone~master~I928fcb8943319e5463a246105391a2bafe833a7e","project":"openstack/keystone","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-04-15 20:42:45.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2024-04-15 19:02:09.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2024-02-19 12:42:19.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}},"hashtags":[],"change_id":"I928fcb8943319e5463a246105391a2bafe833a7e","subject":"Enable protection jobs","status":"MERGED","created":"2024-02-16 19:36:06.000000000","updated":"2024-04-15 20:45:32.000000000","submitted":"2024-04-15 20:42:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909238-secure-rbac","meta_rev_id":"7a3987c9d186136196f35364767d84b739e783d1","_number":909238,"virtual_id_number":909238,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-15 20:42:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2024-04-15 19:02:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2024-04-15 16:19:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2024-04-15 19:02:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-16 19:47:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-16 21:08:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-17 01:57:11.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-02-17 01:57:11.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"fd32cddcdc25b2427ce66ae735b0c8d56ef57c79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-02-16 19:36:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06668c322207b46ebfb482d1c0e730aebc4b9525","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 19:47:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e7cb8208be84f349a9f78f4e035e2f6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5bdd4cca266949bfbebd015f0e6b95be : SUCCESS in 10m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9a454c4ebe4c4e93b6d2862079602566 : SUCCESS in 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b61e623877663813367e58f2b6c9123d9ee36b38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 21:08:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e3cf24fe85b45be88b21f17e5276931\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a25dade568f9471ab8d19e16da47f068 : SUCCESS in 19m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3469ea0f4c65417b8b33f9eeb64ac55d : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7710843b0f6440789de1f73335939e3 : SUCCESS in 11m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5717be6ab9446b09983d92ec9079434 : SUCCESS in 11m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c5d7d3c58bf4d26a4b59a55f09a7c99 : SUCCESS in 12m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/dc39b4eb0f234bf5b03e36a7f3cc851f : SUCCESS in 1h 05m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35c937b3b66e4d14ba730cf655a8d88b : SUCCESS in 1h 26m 57s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7ca902b9908a48e5be74a26ec3c177cb : SUCCESS in 12m 35s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/70003960339044d5a412786afcc3191f : SUCCESS in 35m 47s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/5362ca8a4850454eb699ce07e3df733e : SUCCESS in 41m 30s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e27068fd0a14ea78403878482f3e493 : FAILURE in 35m 13s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/3621d8b990ab485d8396e08aad7e06bb : SUCCESS in 35m 03s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c6e8dcab75c74026be0a021656595232 : SUCCESS in 13m 04s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c4d5bb42e993410cbe7645230a70b121 : FAILURE in 23m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6eac4dfee1046d29276cc2b2cd282ec : SUCCESS in 55m 47s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/03802ca9dbb44f5aadea5e02eb86e3e0 : SUCCESS in 42m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e87019022e30365dc4b3e59f6ceee9ca83dabd40","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-02-17 01:56:52.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"4064f852276308df7580253c7809e8393bf94a2a","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-02-19 12:42:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42a8de5ff77fefded924a8ae3dc42919695d2966","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-03-22 19:29:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f45452b852fd3cfce5b387ce1c599d8f5e6df7da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 20:07:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20e8afd118844e23b4e396d93f0b38c6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9e1daf7155c949a3902ec6956a30c52e : SUCCESS in 30m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/580ae899ddaf4e018f03d12bfebc37f3 : SUCCESS in 35m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/14147b6ae0134aadbf67de25d1e3e53e : SUCCESS in 13m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4431ab82282d47ab5d80a24debe203b05d1643f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 21:15:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/510ca756fc9b4cf59f68a2c2e3a00df7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2a6a3d75be84e3a8e66b4d1f72b8869 : SUCCESS in 13m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e15df8f82c1047f79db885ae99906116 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb3c47f7e57d4401b9e14e49ff58f1c8 : SUCCESS in 10m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/35630828e27640e8b2e83f539dfaceb5 : SUCCESS in 10m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7d13a3fd4aa94fd1979bb0d880503e96 : FAILURE in 14m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d39000855324042a743653334a3d5a8 : SUCCESS in 12m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/12a3160a6c054315b3e3d53fa087e0dd : SUCCESS in 42m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b00cb2c5883b4abf97db58af466a7122 : SUCCESS in 1h 40m 19s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/00c82030b1f142e5b74e57c79c2fb090 : SUCCESS in 13m 28s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/9ac572a414984a5dbc5e7320b70637f9 : SUCCESS in 32m 11s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/b58b14a52afd4a368d9557bb96584961 : SUCCESS in 24m 57s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2bb01617dbed45a79f255484bd51c6ee : FAILURE in 26m 37s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/6d38eeca0045408791b8819567a63418 : SUCCESS in 18m 24s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/f1f569860c84425496a377d6dff4b0ca : SUCCESS in 20m 32s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/aa482b3aa2ee43c9a8d1304122aeb248 : FAILURE in 10m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f42d198f5da45bd9b972ca59691bc59 : SUCCESS in 34m 01s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/a744bbabd06240d4b44e92015ad6078c : FAILURE in 26m 21s","accounts_in_message":[],"_revision_number":2},{"id":"3643377dfc706755c5cf186bb30167f14803c820","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-03-27 18:26:26.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2d1664d21089ff4f9f6adcdcb34ca6d689a8eddc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 19:02:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea84f9ca8be34caa8a0205800dae14fe\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c7ef61bca5b241dc82a7e1ebe4f6c9e6 : SUCCESS in 11m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5d95c3d4a2aa4e19bdcddcc1ae8a0d68 : SUCCESS in 34m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"be641238500aa635727d7d2c240997ce4cfa3c4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 19:42:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca89053a148b4f14a3e88894b6376258\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d5a0fa638864e98af75f792c83662ce : SUCCESS in 14m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4bc27aceec44d718c7deead76e0848a : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cecf8159b6ba464fbfbb80e6137dd086 : SUCCESS in 12m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/29ad854eafc84628ba2f6ab82bb03774 : SUCCESS in 11m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5a26eb3b64d0403b960af3dd91b1b768 : FAILURE in 14m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6da63e6f11024d6cb84ece29c1b35494 : SUCCESS in 11m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/129b13b3508c43a4bcbdd8a1b5c50714 : SUCCESS in 1h 07m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8fab79124ab241e78ef4b2a8f7f27221 : SUCCESS in 1h 13m 22s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7ed87d28cc6a4256be7b2971378c09e3 : SUCCESS in 14m 39s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/5bd28285197e442b9bd57f2448b5982c : SUCCESS in 22m 07s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/98ed4d6982a940f1b579e6c025a2b2f3 : SUCCESS in 30m 10s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c674545b5204d15af92d6d86877449e : FAILURE in 16m 42s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/83f40819456c4f699cf92e05702608eb : SUCCESS in 22m 53s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/504292bb537c44eaa010828e3eba065b : SUCCESS in 13m 41s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/421477bea4974caeb59e1407daf28407 : FAILURE in 23m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec87bea3d1ec48e4a07de9260e5a9315 : SUCCESS in 43m 12s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/d6a26ac30ec44236ac941a7774987419 : FAILURE in 38m 13s","accounts_in_message":[],"_revision_number":3},{"id":"d6cea5cddbd471f24d83d448002bd218f80b3b52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-03-29 15:33:22.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"8b70e32020b4f7a086e67b9506aa7873b3290294","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 15:48:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfa9cc674efd4cef928b60b56fc533d8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e3e8154479174aad90c5ad6030d77396 : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7cc17f220c2e41219093d2a201e533b0 : SUCCESS in 13m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ec0931ab11a7a3726014b7d97f5e73bfe95ae268","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 16:54:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9af237994e1f40ed905dd67f6e435335\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a4c7a55e14e4ccd9fb2e4a28be3110b : SUCCESS in 15m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e33246cd89fa4e219038731ccb3ba300 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19844f2bcca44a8f9fbdea3b4fb8b262 : SUCCESS in 13m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/569ab41f09d144748dc711690e38cf79 : SUCCESS in 10m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e9a3e9ae17f04ced8a21b14ec2051cb4 : FAILURE in 9m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c397ed4f5b3f468a9784b28f2d1e0f8b : SUCCESS in 11m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/105e6d6879d744cd884c5320e381b379 : SUCCESS in 1h 11m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d18b5c12917d4194b8a6d86fa956ef37 : SUCCESS in 1h 18m 36s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/08fb0063ae5940c88074023289e4b188 : SUCCESS in 15m 30s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/42261f2cd6784e86803cea607f9d491a : SUCCESS in 32m 06s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/5e02567ba48943b7afcc1446b04dac9a : SUCCESS in 27m 30s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91ba9c66df9345f794394eae3946a413 : FAILURE in 28m 16s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/64f24f60f1af4b9d87864584781afbb4 : SUCCESS in 34m 33s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c4c850513b1a4907ad97b973b17dfdd0 : SUCCESS in 10m 53s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/0e5ce73022214baebd946864867b29e7 : FAILURE in 12m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c574928dac674cf0b908a01007d7041d : SUCCESS in 43m 54s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/8f735609e62f41de9aa73970712d677f : FAILURE in 32m 46s","accounts_in_message":[],"_revision_number":4},{"id":"30b176786395cea3a6ce900792619994ae3890ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-04-03 02:04:43.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"0f4b89fdfd13024d899895fd853e25dfa220c387","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 02:27:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73a924d770fe4634ba2267bf4990a19d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/67115d534d314f44a9981044f1a4d07f : SUCCESS in 11m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/772143a60cea4ca0b5b6648cf0cc721a : SUCCESS in 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e20b3b7475073107413a31fa7d298e5bd176d8dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 04:01:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7705adbac584c6ca032c8453fbb84aa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b7217ebd39c4025824d3c3935c72b27 : SUCCESS in 13m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b8378a3a50b4d00b43b81fc51a2c9dc : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04813b8ae05545f986ffe9318f7c8a15 : SUCCESS in 11m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3666bfb22dce476abb12f1fec6240f44 : SUCCESS in 10m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3cd4bfc98aa34e71944439ce81631f71 : FAILURE in 11m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa2bded32fb240dfaaca63cb657d7367 : SUCCESS in 12m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/c5fd32261d56473392ed6ab49e27a551 : SUCCESS in 1h 02m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e4d1d3318934512a06988f6d036aade : SUCCESS in 1h 28m 37s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a8d92e93b2f14df0b3964231a4ae8f4d : SUCCESS in 13m 45s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f056ed100bcd409b9bbfad8ba7a1874f : SUCCESS in 33m 22s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/2aa0ba1fc62b4368a8e1d0adee96428c : SUCCESS in 47m 57s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62995a42374a478f9c6957f733ef12be : FAILURE in 28m 00s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/bbe1c3b2bc56492f8772d140ad76ee9b : SUCCESS in 32m 03s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/33eb9eed8cf849d99bec1b5daa0b6723 : SUCCESS in 19m 40s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ec01be4f932a4832b4dafae5c95ae433 : FAILURE in 26m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67d9b94e13e7422b9b99bf3649de73ed : SUCCESS in 59m 40s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/de3194105c2340eb95e336d6123e735b : FAILURE in 44m 30s","accounts_in_message":[],"_revision_number":5},{"id":"c1227eb9e84e0bc9c8d61d7edbcab2ca00bd5a54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-04-03 04:18:29.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"7dd5bb53829c7589710be2dbabf8b40c7264131f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 04:23:17.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/914934 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"0943475297703d4285f01ff4717cc193a5a4b9c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 04:25:54.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/914934 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"8a9cf2458f22a1002c36922e2324d3d6f4db73a2","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-04-03 18:51:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d9cbf5af98e5166183cda84d0f74bed5af30e054","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 20:08:30.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27c3e98fa382469ca83836b00ff338a8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a2ef65071114134bfb00e07178f0bc3 : SUCCESS in 12m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/74f6818d5ad24538bd04befb90190094 : SUCCESS in 55m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"dbdf47cf2a4b3cea57d48a406d1bb19ed5ec9ff0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 20:24:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c9eec9193a6417ead9d3809dfc26991\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adc5790a79f24dfca354d39234e33dc4 : SUCCESS in 20m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b7d0e7ccce648ac8a783f733cce1dec : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fee3d9229b74a4c99f9d4026ee3d4bb : SUCCESS in 12m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ed724728d7648f8972460020cc619f4 : SUCCESS in 12m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e9b4fee0a0047278b7b3536f63ac511 : FAILURE in 17m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4553408cb614b12b6d83751f19f849e : SUCCESS in 12m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/532ccea2b35c43a6930738d81197dc67 : SUCCESS in 1h 02m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9e4c796b7804361ac25b5f330f8dea0 : SUCCESS in 1h 06m 14s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ad6e0ad928e541e48bb9372921c2394d : SUCCESS in 15m 03s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/3c2a67fe78cf47e1b245ca1e5aaada98 : SUCCESS in 22m 34s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/89fd97a16ec24e5fbdd10a4947b10a81 : SUCCESS in 46m 21s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2422767c223248708415fe33f88e8484 : FAILURE in 31m 09s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/a1912b13774e43469842582d2778a170 : SUCCESS in 39m 28s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/58cc8602b1434d0fb7e6fcc7c49d2335 : SUCCESS in 21m 57s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/6f5d293529734a979d5159ad28ab4fed : FAILURE in 12m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca18c5c7e49b4b83ab03a8c788b634c9 : SUCCESS in 1h 02m 09s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/b21ccedcfe984a729c4dff0afe1d32e0 : SUCCESS in 47m 46s","accounts_in_message":[],"_revision_number":6},{"id":"b45434704383995f4847a4af8e9c16fc9abf4404","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-04-04 13:26:29.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"1d6d0b661e7b7cabaf8f31e7ba226f068639b8ea","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-04-04 13:28:10.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"9c036ce765f14ffe8f13a546c8ceebd536a261e2","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-04-04 18:41:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cb605160ba7639e174c9e46b42d1b78c25118bb1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 18:54:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6744dc56aab47809c0f82a2db9775b7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/03d4c38a14134accb3cbea717b981cda : SUCCESS in 9m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e9884d250e2a4f2b9a7a3f90f7062eb1 : SUCCESS in 12m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c9a1f0592e6633f177733b1fddca25cbb1e7ef81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 20:16:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/552fedb4772842b1a9b6bbcd039404e7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab9e3cb18c7a49049da68af71caac7e4 : SUCCESS in 13m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7233d120b6a4eb4988525e129b94907 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e3d2413c90949138cbdf6022c504515 : SUCCESS in 11m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c354aa339a764dbb8dc78ed0aee3d042 : SUCCESS in 11m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4c2e814aa525400a856fd8cdd472d2c9 : FAILURE in 9m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50a8c67e346a4ac3ae4c18b7e4cf49ee : SUCCESS in 11m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/72f659f92cfa455d88c8e8fd11736c69 : SUCCESS in 52m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22eaf382477f4a31b37c3e661ad71abc : SUCCESS in 1h 17m 27s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bceae1e13ab74e3d8280db6a6db9508d : SUCCESS in 12m 10s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/c471f02dfc6b465f81689ef02a3eeb83 : SUCCESS in 33m 56s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/5f168a764eb14320b5b4f2520ba69958 : SUCCESS in 30m 45s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/055b127473bf4ea5892747a6568ec164 : FAILURE in 14m 38s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/6a0cc1bbb578455082b478984aa522bd : SUCCESS in 32m 27s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/acbf94626f8147d884a58b0ad6074b38 : SUCCESS in 10m 03s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/a851cfec2de64b649713c084ad956c9c : FAILURE in 24m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d54caf3fee3b42e1967eff878eae3ee4 : SUCCESS in 41m 12s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/a56326ccecb04ed88070c357f8854d45 : SUCCESS in 34m 57s","accounts_in_message":[],"_revision_number":6},{"id":"fffd7226fe531ddee2e4b3980a1a500bb2fbe37f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 15:10:31.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"15a623118e9fae52088f1dbdd79aa78f1e5e6f90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 15:55:05.000000000","message":"Patch Set 6: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/914934 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"c0bad8ad8d7cd88c8aab6f17196f2e7888f1cf8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-04-15 16:19:26.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":7},{"id":"eaecaf2b6009ac4a21f6ea8051cd03270444f882","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 17:28:05.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa297949ee08427ba186fd8b73224fa1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/690fc8c5ed6d4d489df1577e7db21474 : TIMED_OUT in 1h 01m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0e58e5eb84e041a68257359df2e91449 : SUCCESS in 26m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"10365e8558d6c35ecb1131ca4908709da335b366","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 17:51:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54dfc1d5a1c3413ca2e99ac8c5a704d1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a3f5f0041c9437f8769335cf0d8688c : SUCCESS in 13m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/654e94de9e1b439eb248248a05dfd929 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5174c39312d459cb3aa33bd9b62d1c8 : SUCCESS in 11m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/308eea29027c4726ae0912d2c669adc3 : SUCCESS in 8m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7c031a6992f84d908efd4bf5b1d5282a : FAILURE in 10m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db6a18b22b794d9cad63e38b4c9eb868 : SUCCESS in 11m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/4aaadf11a5b5497284c039cb1debb42f : SUCCESS in 1h 01m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dcc9a35e95444893b33a28e46c9b6b47 : SUCCESS in 1h 25m 03s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/413502c793aa4488a5caff5792a8edca : SUCCESS in 13m 40s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/9c9baef001b74a2995a915067fc2bdef : SUCCESS in 33m 06s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/b1302b180da043a88ca86b45cef82e8a : SUCCESS in 37m 06s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a76ecfede4ee416cb8bc67717d5f8baa : FAILURE in 30m 58s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/a5676966f330447c93eee90d16ebf908 : SUCCESS in 25m 31s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/61ddb8efb3694aac84bc596a6381a5aa : SUCCESS in 10m 47s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/7ca679fca76c4b9a9911663043bb2f6e : FAILURE in 25m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/164b4b2060c844bdbd61dd3ba917cc4c : SUCCESS in 1h 06m 36s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/aef8a6240eda4128bb92cf67308866e6 : SUCCESS in 30m 45s","accounts_in_message":[],"_revision_number":7},{"id":"134a86b480f8000924e87c7eeacce014aa04510c","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-04-15 19:02:09.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"2f8482866606da64a4caa10b96dfba78c5bf7127","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 19:02:50.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"8db827da2e9776d831469b658bc31c2795aa4eda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 20:42:45.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a4dbae1e363415ba79425dff060c3ae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2392dbb0e26d4158bda88b5a8ee13475 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3689e02088d441edaaeb949abc6f639b : SUCCESS in 11m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6454ea525d3843f38407abdbf57370c0 : SUCCESS in 10m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86fcdb544f8741d99de9af1be849e68d : SUCCESS in 12m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/c8dd73fd17ae4b05aecf9ea9cd80843d : SUCCESS in 59m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3df8944bf1ee4107ae1808a1773f3fa9 : SUCCESS in 1h 34m 47s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/58cbf485efd94d36bd67d33c1464bbd0 : SUCCESS in 13m 41s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/007ef4ca9afa4d61b4f65437189df1fe : SUCCESS in 22m 27s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/6080ea35c7c94435ae3a37eca7c32f0d : SUCCESS in 33m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c5ab54e20104c00a95a117fa7299287 : SUCCESS in 1h 01m 46s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/768522953f66446aa9a5927e9a604487 : SUCCESS in 29m 07s","accounts_in_message":[],"_revision_number":7},{"id":"6a504033c458faca497033df622154d06491fffc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 20:42:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"7a3987c9d186136196f35364767d84b739e783d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 20:45:32.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33c0616b874d458bb28a8a1713830962\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed27a841c4f14da6814cd67a361926b8 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"7a6e1a0bdc79927e5d7fd6ad7e6dda2e04c8342c","revisions":{"b7986a9dd5577b8c132ae17e7e1dfe7791769327":{"kind":"REWORK","_number":1,"created":"2024-02-16 19:36:06.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/38/909238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/38/909238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/38/909238/1"}}},"commit":{"parents":[{"commit":"7dc175a41f92e3f01cf26912431d0f2c98a03b32","subject":"Normalize policy checks for domain-scoped tokens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7dc175a41f92e3f01cf26912431d0f2c98a03b32"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-02-16 19:35:17.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-02-16 19:35:17.000000000","tz":-300},"subject":"Enable protection jobs","message":"Enable protection jobs\n\nThis patch re-enables the protection gate jobs now that policy changes\nand tempest test changes have merged.\n\nChange-Id: I928fcb8943319e5463a246105391a2bafe833a7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b7986a9dd5577b8c132ae17e7e1dfe7791769327"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b7986a9dd5577b8c132ae17e7e1dfe7791769327"}]},"branch":"refs/heads/master"},"6d45d6185d0e0beea01f23839231e521e786ec35":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-03-22 19:29:26.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/38/909238/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/38/909238/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/38/909238/2"}}},"commit":{"parents":[{"commit":"39f9048244b0227dd6aea4fc3a385d4e22b463f7","subject":"Run Secure RBAC tests as project-admin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/39f9048244b0227dd6aea4fc3a385d4e22b463f7"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-02-16 19:35:17.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-03-22 19:29:26.000000000","tz":0},"subject":"Enable protection jobs","message":"Enable protection jobs\n\nThis patch re-enables the protection gate jobs now that policy changes\nand tempest test changes have merged.\n\nChange-Id: I928fcb8943319e5463a246105391a2bafe833a7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6d45d6185d0e0beea01f23839231e521e786ec35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6d45d6185d0e0beea01f23839231e521e786ec35"}]},"branch":"refs/heads/master","description":"Rebase"},"d14c5458b2499eb03288d4679775be0adf77b10b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-03-27 18:26:26.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/38/909238/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/38/909238/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/38/909238/3"}}},"commit":{"parents":[{"commit":"5849bf808b22c57e81c0da8fa121ddfd2c92c101","subject":"Allow admin to access tokens and credentials","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5849bf808b22c57e81c0da8fa121ddfd2c92c101"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-02-16 19:35:17.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-03-27 18:26:26.000000000","tz":0},"subject":"Enable protection jobs","message":"Enable protection jobs\n\nThis patch re-enables the protection gate jobs now that policy changes\nand tempest test changes have merged.\n\nChange-Id: I928fcb8943319e5463a246105391a2bafe833a7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d14c5458b2499eb03288d4679775be0adf77b10b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d14c5458b2499eb03288d4679775be0adf77b10b"}]},"branch":"refs/heads/master","description":"Rebase"},"9ba38540a87c59d02c3fa79e34fb2f06b001a259":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-03-29 15:33:22.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/38/909238/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/38/909238/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/38/909238/4"}}},"commit":{"parents":[{"commit":"5849bf808b22c57e81c0da8fa121ddfd2c92c101","subject":"Allow admin to access tokens and credentials","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5849bf808b22c57e81c0da8fa121ddfd2c92c101"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-02-16 19:35:17.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-03-29 15:33:22.000000000","tz":0},"subject":"Enable protection jobs","message":"Enable protection jobs\n\nThis patch re-enables the protection gate jobs now that policy changes\nand tempest test changes have merged.\n\nDepends-On: https://review.opendev.org/c/openstack/keystone-tempest-plugin/+/914558\nChange-Id: I928fcb8943319e5463a246105391a2bafe833a7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9ba38540a87c59d02c3fa79e34fb2f06b001a259"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9ba38540a87c59d02c3fa79e34fb2f06b001a259"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f9497a0bb7c12772d3454b4c8d12863ced1aa15c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-04-03 02:04:43.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/38/909238/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/38/909238/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/38/909238/5"}}},"commit":{"parents":[{"commit":"b31007e1b2ecbea5e1268d3e28d6230d0f5d09b2","subject":"Allow admin to access tokens and credentials","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b31007e1b2ecbea5e1268d3e28d6230d0f5d09b2"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-02-16 19:35:17.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-04-03 02:04:43.000000000","tz":0},"subject":"Enable protection jobs","message":"Enable protection jobs\n\nThis patch re-enables the protection gate jobs now that policy changes\nand tempest test changes have merged.\n\nDepends-On: https://review.opendev.org/c/openstack/keystone-tempest-plugin/+/914558\nChange-Id: I928fcb8943319e5463a246105391a2bafe833a7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f9497a0bb7c12772d3454b4c8d12863ced1aa15c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f9497a0bb7c12772d3454b4c8d12863ced1aa15c"}]},"branch":"refs/heads/master","description":"Rebase"},"7687e1f38e63e24e816538aa6f12d68a3bdfa00b":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2024-04-03 04:18:29.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/38/909238/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/38/909238/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/38/909238/6"}}},"commit":{"parents":[{"commit":"b31007e1b2ecbea5e1268d3e28d6230d0f5d09b2","subject":"Allow admin to access tokens and credentials","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b31007e1b2ecbea5e1268d3e28d6230d0f5d09b2"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-02-16 19:35:17.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-04-03 04:18:29.000000000","tz":0},"subject":"Enable protection jobs","message":"Enable protection jobs\n\nThis patch re-enables the protection gate jobs now that policy changes\nand tempest test changes have merged.\n\nDepends-On: https://review.opendev.org/c/openstack/keystone-tempest-plugin/+/914934\nChange-Id: I928fcb8943319e5463a246105391a2bafe833a7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7687e1f38e63e24e816538aa6f12d68a3bdfa00b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7687e1f38e63e24e816538aa6f12d68a3bdfa00b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7a6e1a0bdc79927e5d7fd6ad7e6dda2e04c8342c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-04-15 16:19:26.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/38/909238/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/38/909238/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/909238/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/38/909238/7"}}},"commit":{"parents":[{"commit":"8ca73f758bb613a57815fbe4ae78e3d2afa4af49","subject":"Merge \"sql: Fixup for invalid unique constraint on external_id in access_rule table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8ca73f758bb613a57815fbe4ae78e3d2afa4af49"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-02-16 19:35:17.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-04-15 16:19:26.000000000","tz":0},"subject":"Enable protection jobs","message":"Enable protection jobs\n\nThis patch re-enables the protection gate jobs now that policy changes\nand tempest test changes have merged.\n\nDepends-On: https://review.opendev.org/c/openstack/keystone-tempest-plugin/+/914934\nChange-Id: I928fcb8943319e5463a246105391a2bafe833a7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7a6e1a0bdc79927e5d7fd6ad7e6dda2e04c8342c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7a6e1a0bdc79927e5d7fd6ad7e6dda2e04c8342c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
