)]}'
{"id":"openstack%2Fkeystone~909769","triplet_id":"openstack%2Fkeystone~master~Ia94b4e7151410515a3ecf99185042dae82bf1b7d","project":"openstack/keystone","branch":"master","topic":"drop-opensuse","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-03-08 17:05:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia94b4e7151410515a3ecf99185042dae82bf1b7d","subject":"Drop keystone-dsvm-functional-federation-opensuse15 jobs","status":"MERGED","created":"2024-02-21 17:09:00.000000000","updated":"2024-03-08 17:05:45.000000000","submitted":"2024-03-08 17:05:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909769-drop-opensuse","meta_rev_id":"73c17cafb018229de88e354c94be2262b448439d","_number":909769,"virtual_id_number":909769,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-08 17:05:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2024-03-08 15:23:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2024-03-08 15:21:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":1,"date":"2024-02-22 03:46:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2024-03-08 15:23:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-21 17:20:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-21 19:13:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-22 03:46:30.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-03-08 15:21:30.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-03-08 15:23:42.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"a2cd80b62b2b2334a8e3696a455b0a4927e0cf66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-02-21 17:09:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c7109a8e1a7832f90d8f31fbc6065f1a6c7fcd5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 17:20:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd1b598ff0194511bfa6be22b3c555c1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9598357959644619b5b45d042f668e1 : SUCCESS in 10m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f2b2ca3ec9804722a132b02ebf231667 : SUCCESS in 10m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da0f833c231b20e14ccdc235b1aae4ce86977778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 19:13:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22dc424e50ae4bdc9f941b32b04c7c0d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/730e38880dcb438a9b80980ae5278f50 : SUCCESS in 14m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/836ff9a883534acebc67b873dc7d65c9 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2453118bc63548c8a4e6d21ef4ea662d : SUCCESS in 12m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7be4b9d004e145249fb133861af1b587 : SUCCESS in 10m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9124d93325614906adf4ec674f098181 : SUCCESS in 13m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/a62ba3e20f744c798caaea8adddb5109 : SUCCESS in 1h 06m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8399702bdfe14b7baef056bcb3627e8d : SUCCESS in 1h 55m 33s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1bdac6d5f7ca4f579ca4c21e78a28620 : SUCCESS in 13m 04s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/599678c7fc2e4cc4ba62a3dbddd554d9 : SUCCESS in 30m 03s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/4181d558bbfe47e394fb39a6f918dc22 : SUCCESS in 30m 35s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1269cc2a1ceb4147a8037999f1d1f1be : FAILURE in 38m 35s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/5aa9e35434c441b9b16b45bb1e6c21e1 : SUCCESS in 33m 43s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/b4f4f0ff0d37453d8c3831496d39dce8 : SUCCESS in 21m 43s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/11668d28f63e4545933820881b724222 : FAILURE in 23m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/986f3db1b8c7486887587a07dca0a043 : SUCCESS in 47m 37s","accounts_in_message":[],"_revision_number":1},{"id":"d7d5bfa330b306af2241783dafe2a23512704db4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-22 03:46:30.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74bf56312050a90700fa622733599874b289fb20","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-03-08 15:21:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c774fc825c445cb40bebe290a68d041895ddf153","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-03-08 15:23:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"80fe786bdad28c49805751a8341ac3968a967a00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 15:24:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0dbee2ba4ff8c4a691987464b49f1f4f0816cd67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 17:05:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5a5567f86df462fa2c52094d980dc61\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28fc5bf64148404bb672d48767ab623c : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f31c8557fb745e5a0980d19edd3b8aa : SUCCESS in 10m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4ce27f3011d2420cb7f8d054f4d98d48 : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3551acb88b24522b72c2a65c1bc5d51 : SUCCESS in 11m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/eabe8089136e4e89b7561ecfa7e4736b : SUCCESS in 1h 05m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5166f467e1a415681a3607d0bcc56b5 : SUCCESS in 1h 34m 01s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9d9feca0e23f43a780abe49fdb424387 : SUCCESS in 13m 49s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/19fa5995833b49f9a807feee589bfb0b : SUCCESS in 23m 14s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/7064ff64faba45a3ab9b84643d2109de : SUCCESS in 34m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64ddddb8515b412da0ca666d8ea19707 : SUCCESS in 1h 12m 10s","accounts_in_message":[],"_revision_number":1},{"id":"73c17cafb018229de88e354c94be2262b448439d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 17:05:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da1c884b90c2f45c131bc45cd06e90e5f83e70fe","revisions":{"da1c884b90c2f45c131bc45cd06e90e5f83e70fe":{"kind":"REWORK","_number":1,"created":"2024-02-21 17:09:00.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/69/909769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/69/909769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/69/909769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/69/909769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/69/909769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/69/909769/1"}}},"commit":{"parents":[{"commit":"7dc175a41f92e3f01cf26912431d0f2c98a03b32","subject":"Normalize policy checks for domain-scoped tokens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7dc175a41f92e3f01cf26912431d0f2c98a03b32"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-02-21 17:08:05.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-02-21 17:08:58.000000000","tz":-480},"subject":"Drop keystone-dsvm-functional-federation-opensuse15 jobs","message":"Drop keystone-dsvm-functional-federation-opensuse15 jobs\n\nThe OpenDev team is planning to remove OpenSUSE LEAP 15 images as our\nnode builds and mirrors are for 15.2 which is ancient and no one is\ncurrently working to modernize these test environments. On top of that\nLEAP is apparently going away in the future and will be replaced with\nanother distro.\n\nChange-Id: Ia94b4e7151410515a3ecf99185042dae82bf1b7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/da1c884b90c2f45c131bc45cd06e90e5f83e70fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/da1c884b90c2f45c131bc45cd06e90e5f83e70fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
