)]}'
{"id":"openstack%2Fkeystone~912074","triplet_id":"openstack%2Fkeystone~stable%2F2023.2~I5a93d0cc179468cfded5c939f2ba97b0def445ad","project":"openstack/keystone","branch":"stable/2023.2","hashtags":[],"change_id":"I5a93d0cc179468cfded5c939f2ba97b0def445ad","subject":"Fix old arm64 job template","status":"ABANDONED","created":"2024-03-12 17:31:27.000000000","updated":"2025-04-28 16:01:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":908851,"cherry_pick_of_patch_set":1,"meta_rev_id":"4a637d012371aa0c46502a44cf3fe34417136868","_number":912074,"virtual_id_number":912074,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-03-12 19:06:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-12 17:45:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-12 19:06:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-28 16:01:29.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"11daa5a13ef09efccb35360abb172c3b68e280b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-12 17:31:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"63e29e16a2aa711060e1be958ffe6f69fb40f5f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 17:45:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8847eade4484e2ca984c12b26deda1b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4978aab3a9324bec88de7c209b336343 : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/de70f1714227433fbbb1006b9a693cba : SUCCESS in 12m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/653528b577e24c9f81b5219cc36ae820 : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8c4f2ee832bd414ca6a9cc5a254f3189 : RETRY_LIMIT in 1m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca9f70426c45dae48173026bcd484c18bbfc7f40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 19:06:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d07fe4b8d635419ba162734e9777a957\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/885aeb91bc264fb8ba7272115801cacf : SUCCESS in 13m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88f9000df1344771b2944e84b5cac3c3 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae862ae68f3647e6b5ce4439a17104d7 : SUCCESS in 12m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21e2462016aa4faabfb18f300b68a6fd : SUCCESS in 13m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7494b1b2d0f44ed8afbf46ac26654fd6 : SUCCESS in 12m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/51d75bb8bef94ede9436e7585e744fe2 : RETRY_LIMIT in 2m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/097c6d90d03e4ff3bf07b832480ecef9 : SUCCESS in 11m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/942d22aaac6f43e5ad7b1f134968a248 : SUCCESS in 49m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e59c66ae8e9441bd89ccb1a23a78af1d : SUCCESS in 1h 28m 47s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/cccfeb4d91804d51abf354ee46138fea : SUCCESS in 33m 11s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/3d7f13d68b294c449bd42695e34fa943 : SUCCESS in 29m 22s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56a0cdf5e448450a99cbaa32f4ddd79a : FAILURE in 19m 54s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/ad845effb0884162bcb70f27f2812ac8 : SUCCESS in 30m 21s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/477feb2c1201476b9057d6efd861fc49 : SUCCESS in 22m 02s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/cca1e456c982470f9f3d5d6b56481dda : FAILURE in 13m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04e2c5fef6264c8fa69d5fb3696c6971 : SUCCESS in 59m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4a637d012371aa0c46502a44cf3fe34417136868","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-04-28 16:01:29.000000000","message":"Abandoned\n\nstable/2023.2 branch of openstack/keystone transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90da18ce4a41f7524291a0a794390786716db649","revisions":{"90da18ce4a41f7524291a0a794390786716db649":{"kind":"REWORK","_number":1,"created":"2024-03-12 17:31:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/912074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/74/912074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/74/912074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/74/912074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/74/912074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/74/912074/1"}}},"commit":{"parents":[{"commit":"5a55e9de15c7f390e43addc5f3ff1a4809ec1a5b","subject":"Normalize policy checks for domain-scoped tokens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5a55e9de15c7f390e43addc5f3ff1a4809ec1a5b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-12 17:48:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-12 17:31:27.000000000","tz":0},"subject":"Fix old arm64 job template","message":"Fix old arm64 job template\n\nChange-Id: I5a93d0cc179468cfded5c939f2ba97b0def445ad\n(cherry picked from commit ae765c33ee6cd8368d21d925eeacbcca49ca7ea9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/90da18ce4a41f7524291a0a794390786716db649"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/90da18ce4a41f7524291a0a794390786716db649"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
