)]}'
{"id":"openstack%2Fkeystone~912077","triplet_id":"openstack%2Fkeystone~stable%2F2023.1~I5a93d0cc179468cfded5c939f2ba97b0def445ad","project":"openstack/keystone","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-18 15:28:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5a93d0cc179468cfded5c939f2ba97b0def445ad","subject":"Fix old arm64 job template","status":"MERGED","created":"2024-03-13 03:56:39.000000000","updated":"2024-03-18 15:31:05.000000000","submitted":"2024-03-18 15:28:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912077","cherry_pick_of_change":912074,"cherry_pick_of_patch_set":1,"meta_rev_id":"e67a8645c6ec0db71a2fac5ea507577979f5cf58","_number":912077,"virtual_id_number":912077,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-18 15:28:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2024-03-18 13:27:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2024-03-18 13:27:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-13 04:11:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-13 05:03:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-18 13:27:15.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"9f09e334f2504ab87db59da144dbe3d7206085a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-13 03:56:39.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"4736a0f7cc6456aa731c5f20af47e4715bc8b915","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 04:11:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/332894c5da364712a8e5f8edd1fb20d8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5851e0fd8c054a318fc25e4f7a23a2ed : SUCCESS in 10m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/21ad7962835e4a0bbc4c6e8fb5f445c9 : SUCCESS in 11m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63893590cffe1fe2914e476913ab9cd252fe0f12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 05:03:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30573e9905fd44658c59f5c832141307\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7691f4c63dd4fa08ee5297ba1d4146c : SUCCESS in 13m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/072f0befef66425194cee8214d0026e1 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92d5dd5859a84a8f99fccc2de07a0ff6 : SUCCESS in 11m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6d38b7852744c1b8b46e1a06f4dbea0 : SUCCESS in 10m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/ef4ad7aee2ac46179970788d6c261af6 : SUCCESS in 49m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5dbe91053ce941eeb8928cc34c64301f : SUCCESS in 1h 04m 32s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f90b44de7fbf43218b05d469a17fa01d : SUCCESS in 23m 07s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/789ff4b0a7ae40d69c26d45bfbfa3077 : SUCCESS in 39m 15s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5a141e0d813406aae34cbb523f68881 : FAILURE in 15m 21s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/7bd5322173554a9e8338327ad0d1332d : SUCCESS in 35m 46s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e36b48bb38184122a974000087214a2b : SUCCESS in 19m 49s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/811491e23cb84622a7efa5890ccfd0c7 : FAILURE in 23m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/615d166dc0624061ae057704d8a88623 : SUCCESS in 40m 53s","accounts_in_message":[],"_revision_number":1},{"id":"c4cd21f1c0963c2428456003e0890aafa8576676","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-03-18 13:27:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fbe33930f19ea9d92bf9e6f46d3ea8c9e916c603","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 13:28:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fd5ee5dc8b24a5bbc211932c4ee34eb872ae436c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 15:28:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c32823797454467084b731e73e549b8b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/776071a328894c24b67afb49c14babc5 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01017b1b49064a5680aac32c938bf31d : SUCCESS in 14m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d1784a9529b4259891c89f7c2f6f86e : SUCCESS in 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/ecc7beb726b244a198d75c72ecbc3bc2 : SUCCESS in 1h 07m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/822fd96b8da140029ba3eabae37530e6 : SUCCESS in 1h 15m 43s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/77a3f6a159f042628dc3bd3f5f951312 : SUCCESS in 32m 03s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/3b5915ced71f4c789697029666f031e0 : SUCCESS in 34m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46c47b11c2234453802ce6c115cb702f : SUCCESS in 1h 05m 57s","accounts_in_message":[],"_revision_number":1},{"id":"00cd7453bad5acbdd5f914af20dc177bda87e958","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 15:28:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e67a8645c6ec0db71a2fac5ea507577979f5cf58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 15:31:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c428cd4afe645a99f49f5a241b2938c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e92e84e5741e44d4bf9a627b9957a476 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ec8e7cca791a8854ec4ce436c6a0aa7777993cc","revisions":{"9ec8e7cca791a8854ec4ce436c6a0aa7777993cc":{"kind":"REWORK","_number":1,"created":"2024-03-13 03:56:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/912077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/77/912077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/77/912077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/77/912077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/77/912077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/77/912077/1"}}},"commit":{"parents":[{"commit":"828519f4636a3c4a5c77975b970833e0ae62955c","subject":"Normalize policy checks for domain-scoped tokens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/828519f4636a3c4a5c77975b970833e0ae62955c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-12 17:48:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-13 03:56:39.000000000","tz":0},"subject":"Fix old arm64 job template","message":"Fix old arm64 job template\n\nChange-Id: I5a93d0cc179468cfded5c939f2ba97b0def445ad\n(cherry picked from commit ae765c33ee6cd8368d21d925eeacbcca49ca7ea9)\n(cherry picked from commit 90da18ce4a41f7524291a0a794390786716db649)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9ec8e7cca791a8854ec4ce436c6a0aa7777993cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9ec8e7cca791a8854ec4ce436c6a0aa7777993cc"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
