)]}'
{"id":"openstack%2Fkeystone~912659","triplet_id":"openstack%2Fkeystone~stable%2Fzed~I0a87cc5a35e6033d670bab56d5cdc8b8312819d8","project":"openstack/keystone","branch":"stable/zed","topic":"drop-centos-7","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2024-03-13 16:00:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0a87cc5a35e6033d670bab56d5cdc8b8312819d8","subject":"Remove unused old job templates and experimental jobs","status":"MERGED","created":"2024-03-12 22:21:20.000000000","updated":"2024-03-13 16:05:20.000000000","submitted":"2024-03-13 16:00:32.000000000","submitter":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912659-drop-centos-7","meta_rev_id":"30c7cdedfabddd8ceb00ccf5e801c43bae74f76a","_number":912659,"virtual_id_number":912659,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2024-03-13 16:00:32.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-12 23:48:22.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2024-03-13 16:00:32.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":1,"date":"2024-03-13 16:00:32.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-12 22:21:20.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-03-12 22:33:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-12 23:48:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-13 16:00:32.000000000","updated_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"reviewer":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"state":"REVIEWER"}],"messages":[{"id":"00b44ed6fc1bf4c08ed5a1b8a68313c6335567b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-03-12 22:21:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e863e78b9a5f118f167c3be316f484fb468738df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 22:33:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe2327b1e354488f9276a67ef3888631\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2df68a6c33f746309567eb4829cb4059 : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2e9eb938ef824a369b8baf407f770ad4 : SUCCESS in 11m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a47d906eee78e54804768cd0b369d72371378e14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 23:48:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ae60c1c2a0247d282cafd3ab94953ec\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c9af47f5db44808acc60ba2f0082808 : SUCCESS in 14m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c141ce09347a4464be022df658732de3 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5f4fbb4eeb14ba08266f7eb6b194a55 : SUCCESS in 10m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c60bfdacc8a48f1b06feec14744f50e : SUCCESS in 11m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f97211e11404d27a88eae8787fe37b9 : SUCCESS in 10m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77717b51d5ea46b9aed96719f1bde949 : SUCCESS in 10m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/40331a8146d6462dbf8ef2ef95e0de23 : SUCCESS in 55m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0fad622576854036becd5e9df1b9cb65 : SUCCESS in 1h 21m 40s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2c61390c8c3f48ef994eeb3246896636 : SUCCESS in 35m 26s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/66f131ba4ea447a582e477fbea524261 : SUCCESS in 42m 40s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0057ea1e01414799aac3ecd39d0cac3c : FAILURE in 18m 47s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/2dcce0f9c7294f4c9e0a3888d879eb4c : SUCCESS in 36m 04s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/a00ea6e32707499abb03ca937cff2bc8 : SUCCESS in 26m 10s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/a2db0446a4264c859836fe7412336826 : SUCCESS in 25m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07f49e20c0844dcf99a5596952217a63 : SUCCESS in 43m 29s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/48a6031a68b24a45ad3c1cd0a67fa720 : FAILURE in 40m 45s","accounts_in_message":[],"_revision_number":1},{"id":"256d38efe131bc87df69cf485151d8a5139e14bb","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2024-03-13 16:00:32.000000000","message":"Patch Set 1: Code-Review+2 Verified+2 Workflow+1\n\nBypassing Zuul to merge this trivial cleanup after consulting project leadership.","accounts_in_message":[],"_revision_number":1},{"id":"0377807613208933b47c926a9e24b9f5e761352e","tag":"autogenerated:gerrit:merged","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2024-03-13 16:00:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"30c7cdedfabddd8ceb00ccf5e801c43bae74f76a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 16:05:20.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a4bdb3e6ff84ca59c504acabf83b1aa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4af62cb714f84ab59ac2a69681d42288 : FAILURE in 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f63062d47712406a807ce07b4ff3ec6213b0e824","revisions":{"f63062d47712406a807ce07b4ff3ec6213b0e824":{"kind":"REWORK","_number":1,"created":"2024-03-12 22:21:20.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/59/912659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/59/912659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/59/912659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/59/912659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/59/912659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/59/912659/1"}}},"commit":{"parents":[{"commit":"72a4fc0f3ccf7a5ca9fc40e5364e14f881ec27b2","subject":"Add an option to randomize LDAP urls list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/72a4fc0f3ccf7a5ca9fc40e5364e14f881ec27b2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-12 17:45:28.000000000","tz":540},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-03-12 22:20:58.000000000","tz":0},"subject":"Remove unused old job templates and experimental jobs","message":"Remove unused old job templates and experimental jobs\n\nUbuntu Xenial, CentOS7 and OpenSUSE 15 are all too old.\n\nChange-Id: I0a87cc5a35e6033d670bab56d5cdc8b8312819d8\n(cherry picked from commit 609ec29ac95817941543646d05bf783a3774ea77)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f63062d47712406a807ce07b4ff3ec6213b0e824"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f63062d47712406a807ce07b4ff3ec6213b0e824"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
