)]}'
{"id":"openstack%2Fkeystone~912727","triplet_id":"openstack%2Fkeystone~master~Ie38448c4df404514fc9c65b5a5b48be929b13cc5","project":"openstack/keystone","branch":"master","topic":"closed-branch-regex","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-14 16:45:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie38448c4df404514fc9c65b5a5b48be929b13cc5","subject":"Update regex to detect closed branch","status":"MERGED","created":"2024-03-13 10:27:29.000000000","updated":"2024-03-14 16:50:12.000000000","submitted":"2024-03-14 16:45:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912727-closed-branch-regex","meta_rev_id":"5e8d368ae70313f6712b56d415c057c7743e2302","_number":912727,"virtual_id_number":912727,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-14 16:45:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2024-03-14 13:51:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2024-03-13 20:41:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2024-03-14 13:51:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-13 11:17:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-13 12:10:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-13 20:41:39.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-03-14 13:51:23.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"e20c5271549a49cb92a55fafb053f36970b985c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-13 10:27:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d12a3749da1f0822a160e6ee0b2276dce7ef2f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 11:17:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df09244e93af4cc6ab633164be3fcc37\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb5bdb54f0ef4790b4e3b94271d346f3 : SUCCESS in 48m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/69a4b6c9ca084db5ac6917e1452c5ba0 : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6980066dd4fc45c094bc3a14cf09c94d : SUCCESS in 39m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2046cb2441f2fe618a43c3923125ceccddab42e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 12:10:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47a01ec18ccc4760add1ddb9f818eb65\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa1df1bb166943e9ba091eee1fe4a0a6 : SUCCESS in 12m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5871afbe2fc44f7e994c4dd44e0c66d1 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/200268dee36544fcbd079e4397e36be7 : SUCCESS in 10m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b7b442f9d914e0bb727e200a230b707 : SUCCESS in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5374e2a88c264da19f64cd274bfc2886 : SUCCESS in 12m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/c830648616544386b552f7b1aea95a44 : SUCCESS in 1h 05m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/deb19c79213e409796ae19f51bbdabda : SUCCESS in 1h 36m 52s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c04fd7aaae0e43ecbe38bc123bee8e0b : SUCCESS in 15m 42s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/9e4a582a14664564b4de41e6f65373f6 : SUCCESS in 21m 44s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/7f56b52681234ec3abf0e64f2cd4077c : SUCCESS in 47m 56s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee19c88888fc4d4d83efb8976bc1d17c : FAILURE in 27m 52s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/b1b86b2d432348bfaf5d9925b00637f5 : SUCCESS in 34m 21s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/b85979f5ebb34b96ab24dee5df04ed98 : SUCCESS in 10m 43s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/378841f3e870450a84689346adb84e21 : FAILURE in 25m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bab0ad170ab2480aa268c28024141f25 : SUCCESS in 1h 06m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e24ec09fbb450fdb979f411131bd6a42b3a886a6","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-03-13 20:41:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6454dcbaec0436e91ca69eec478ed746a219b67a","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-03-14 13:51:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b29aaa90229149e4be2f49368a3c050b76d742c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 13:52:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ae1cf628a09205119ca9bff7f88e582bafc00171","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 16:45:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19689e4309324303b9fc4032b3099b9f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86cee8b1f8c14e2582fb0e8ceb398844 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0277eb19a6034ab9bcee53f12b43531b : SUCCESS in 13m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d1602ac7e9814794b772a235e881aca5 : SUCCESS in 11m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ea429609c0945fab73eb22c64ee6df6 : SUCCESS in 13m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/d5aaefe9db20458287e238b5d8f28378 : SUCCESS in 1h 02m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/094bc176da0545c0b98e792a2aea4878 : SUCCESS in 1h 51m 01s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/431f93fe4dae45cfb9b72b88f7959a7b : SUCCESS in 21m 01s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/96027645c63c444bb94c7f19669e5414 : SUCCESS in 30m 08s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/f3734a6278b34d4fb6cf54bd71d4bdf9 : SUCCESS in 33m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee9be5c12bd147d2851066bc7097b5b1 : SUCCESS in 1h 01m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c44ebb90651183275683dc8912d1284cadda8b87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 16:45:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5e8d368ae70313f6712b56d415c057c7743e2302","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 16:50:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c13425ccc935482494913f7e5fc06e53\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c61ad9e25de2473ca52e43b97ce4fb05 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"29fc86390c51b803f8b79f4a2c17ea0f46f66238","revisions":{"29fc86390c51b803f8b79f4a2c17ea0f46f66238":{"kind":"REWORK","_number":1,"created":"2024-03-13 10:27:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/912727/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/27/912727/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/27/912727/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/27/912727/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/27/912727/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/27/912727/1"}}},"commit":{"parents":[{"commit":"5a97b7d847d5471d91b7e41ab0acf65974419c44","subject":"Merge \"Fix policies for groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5a97b7d847d5471d91b7e41ab0acf65974419c44"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-13 10:27:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-13 10:27:16.000000000","tz":540},"subject":"Update regex to detect closed branch","message":"Update regex to detect closed branch\n\n... based on the change made in reno recently[1].\n\n[1] https://review.opendev.org/c/openstack/reno/+/910547\n\nChange-Id: Ie38448c4df404514fc9c65b5a5b48be929b13cc5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/29fc86390c51b803f8b79f4a2c17ea0f46f66238"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/29fc86390c51b803f8b79f4a2c17ea0f46f66238"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
