)]}'
{"id":"openstack%2Fkeystone~913999","triplet_id":"openstack%2Fkeystone~master~I39d50b8e6e55b0835670d753c3783f32b19b6c47","project":"openstack/keystone","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-04-03 09:17:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I39d50b8e6e55b0835670d753c3783f32b19b6c47","subject":"Run Secure RBAC tests as project-admin","status":"MERGED","created":"2024-03-22 16:46:59.000000000","updated":"2024-04-03 09:25:05.000000000","submitted":"2024-04-03 09:17:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913999-secure-rbac","meta_rev_id":"117f7eddb2dcf232fbe0f24a91455faa81cc6509","_number":913999,"virtual_id_number":913999,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-03 09:17:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2024-04-03 01:58:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2024-04-03 01:58:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2024-04-03 07:19:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2024-04-03 07:19:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-22 17:01:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-22 18:25:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-27 14:05:47.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-03-29 07:28:51.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2024-04-03 07:19:46.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"72c6ee6fd1ebf3f4167ee97c956060dfa841e248","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-03-22 16:46:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a9bf864718d814a838653fd94d28cb9e726dd22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 17:01:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/263c7b95c0fe43178f115ed0fc97ade4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f0b4419f268846e49c3edfd09984755a : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c5db7b28f0dc4d5186495c358165c305 : SUCCESS in 10m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/258563e8ff5c4776aedf2e8dc523b88d : SUCCESS in 8m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44776e90832f862a4732dcec79ee13745f280d10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 18:25:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7a81f05f75e4980888e3ec68234a9e7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa057b0361454c5ebd7cb7dea6c1b816 : SUCCESS in 14m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e97c63d205d4c34b8760969373c87d3 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/542fa1b1de874ca59dff376f03f63bdc : SUCCESS in 12m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eaa220bc63e440a089bb77fd61472837 : SUCCESS in 13m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/09eb7766aa1a4668a4b94de92bc3fb12 : FAILURE in 12m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/180505932d8e413e80e078aaef9f46da : SUCCESS in 11m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/bca610b1e4324806bfb654e7ab556641 : SUCCESS in 1h 04m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/20f6d42e648142c983611dcd6dba30ae : SUCCESS in 1h 32m 19s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/14836ca6624642f2a4493fa23eda7576 : SUCCESS in 12m 42s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/39a502f840bc49bbb4f04d57c1a4f664 : SUCCESS in 34m 53s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/ba688fa745494c188c53cde8abe9e49a : SUCCESS in 45m 14s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6de76b986dac457cb41ade5cd4e2ea50 : FAILURE in 27m 31s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/eb60c70b791f4802bb7267229414ca3c : SUCCESS in 33m 46s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/f9c1df8e7af146b9899abaf438443032 : SUCCESS in 19m 34s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/3e73c1f5b98546229a94d337a14ba6f7 : FAILURE in 25m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47ff949bea2c4792bbf10a1226620cdb : SUCCESS in 58m 24s","accounts_in_message":[],"_revision_number":1},{"id":"66a4361820f3b5b2b59c46eedd32ffa5fbb942ff","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-03-27 14:05:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f534da74922cb23bb3d8e3515cb2d7d8817ee41f","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2024-03-29 07:28:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9f1894e2b59b9234b7ecf0dfe77c4111090ffbcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-04-03 01:58:59.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a24fc8f913280529c8f1cc949a97d8d905f9ca38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 02:17:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9caedc2da814f9da41c8944e452c525\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e392cefbceb845df9e7f387ecadc3a31 : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c53f38e97dd64ef29afb75a792e4d2a0 : SUCCESS in 13m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c0c017e4d7e9b61b3030ad34167901d09b31d2cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 03:38:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18b5f07fd19448639846baf8860a74aa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eacf5c011d3b4679bfa3bfdf7d086856 : SUCCESS in 16m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6ef90060a834266b180025c43c03126 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2f89c2c74d1447c889a3080acaecaa6 : SUCCESS in 12m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2587289855df4ab3a754d93208c0aebc : SUCCESS in 13m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f9653da18c33438c86a78bf97e3792ca : FAILURE in 9m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d41c5849559b46d1b9adc697772fe3d5 : SUCCESS in 12m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/f36447d9acb0440fa94492b9febee64d : SUCCESS in 1h 06m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b8e5f56c1dd4749b4ffbb224a947292 : SUCCESS in 1h 33m 00s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/64c45843f9d54a6983f80e57f168586a : SUCCESS in 13m 46s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/fc4b896685be4a9bac2a83ffd4896672 : SUCCESS in 24m 09s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/6ceffe7f0e2e495d8a06c1b8ed26313d : SUCCESS in 44m 25s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f45a8c1903ec42f9ab6dc1dad5b06326 : FAILURE in 16m 35s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/a8e160efa0864a62a433b3d7cb691910 : SUCCESS in 31m 35s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/9b2eff89d88b4cb78d98508be8c33cf7 : SUCCESS in 19m 36s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/2f9d11514db04bf3b4b9d3b3ed72e918 : FAILURE in 24m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fe8aeaba293490f9389386af21f6796 : SUCCESS in 1h 09m 44s","accounts_in_message":[],"_revision_number":2},{"id":"96278efeb77cdb3a6866c698d6f054fba4654779","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-04-03 07:19:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"648033eaf38cbc0a8eddb7dfdde85f9ef0ee2f36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 07:20:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2825919e7696792217bc550a2c3994b07a839c63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 09:17:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89f0e0968235423ab5449e496054df3c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0105d65e38f34b5e9dfe0b91d2f739ac : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d3ea4077c83477eb1603fc1672ca19e : SUCCESS in 12m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bdbb80332a1e4f91aaa4f0e9b2630533 : SUCCESS in 10m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c19d695678140db817db8896c735baf : SUCCESS in 12m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/59cdcc0a0c9349bbb69c161fb201a559 : SUCCESS in 55m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8007576c3f9645079d9c618b974f88d3 : SUCCESS in 1h 31m 53s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b99060b606154a2487237212f044b482 : SUCCESS in 14m 20s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/34cfb33c9c764e4ea7aa41c6c1f491b8 : SUCCESS in 24m 27s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/069f694179494eaaa358bd76408b168c : SUCCESS in 17m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8f6c1c6c70f4133826274255904ce7b : SUCCESS in 37m 25s","accounts_in_message":[],"_revision_number":2},{"id":"2c9b58a93529021c895886a1db5bfdb8ec186fee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 09:17:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"117f7eddb2dcf232fbe0f24a91455faa81cc6509","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 09:25:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec3e4423403343b8ab6441d57c313ae4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29f6be50a2e247e5aad957f00172d46a : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a050129384ac4803d1c56001a3140fc547fe134b","revisions":{"39f9048244b0227dd6aea4fc3a385d4e22b463f7":{"kind":"REWORK","_number":1,"created":"2024-03-22 16:46:59.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/99/913999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/99/913999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/913999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/913999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/913999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/99/913999/1"}}},"commit":{"parents":[{"commit":"4121cf6cb7c3d5e585df827d72e08296664f7326","subject":"Merge \"Add ability to create users and projects from keystone-manage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4121cf6cb7c3d5e585df827d72e08296664f7326"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-03-22 16:41:30.000000000","tz":-240},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-03-22 16:41:30.000000000","tz":-240},"subject":"Run Secure RBAC tests as project-admin","message":"Run Secure RBAC tests as project-admin\n\nThis patch updates the devstack plugin so that tempest.conf is not\nconfigured to use system-admin.  Currently tempest uses an all-in\napproach to configuring admin clients, and forcing system scope in\ntempest when SRBAC is turned on results in test failures for services\nthat don\u0027t understand system scope.\n\nWith this patch, keystone test will be run with a project-scoped admin,\nwhich should be fine since policies have been previously updated to\naccept project-admin tokens as legacy admin for Phase 1. [1]\n\n[1] https://opendev.org/openstack/keystone/commit/f2f1a5c38847ddc5aa28eec9722885d9c64c6e7b\n\nChange-Id: I39d50b8e6e55b0835670d753c3783f32b19b6c47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/39f9048244b0227dd6aea4fc3a385d4e22b463f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/39f9048244b0227dd6aea4fc3a385d4e22b463f7"}]},"branch":"refs/heads/master"},"a050129384ac4803d1c56001a3140fc547fe134b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-03 01:58:59.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/99/913999/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/99/913999/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/913999/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/913999/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/913999/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/99/913999/2"}}},"commit":{"parents":[{"commit":"fc10ccbc8c9798e554add498997535171e0e099f","subject":"Merge \"Replace CRLF by LF\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fc10ccbc8c9798e554add498997535171e0e099f"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-03-22 16:41:30.000000000","tz":-240},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-04-03 01:56:48.000000000","tz":-300},"subject":"Run Secure RBAC tests as project-admin","message":"Run Secure RBAC tests as project-admin\n\nThis patch updates the devstack plugin so that tempest.conf is not\nconfigured to use system-admin.  Currently tempest uses an all-in\napproach to configuring admin clients, and forcing system scope in\ntempest when SRBAC is turned on results in test failures for services\nthat don\u0027t understand system scope.\n\nWith this patch, keystone test will be run with a project-scoped admin,\nwhich should be fine since policies have been previously updated to\naccept project-admin tokens as legacy admin for Phase 1. [1]\n\n[1] https://opendev.org/openstack/keystone/commit/f2f1a5c38847ddc5aa28eec9722885d9c64c6e7b\n\nChange-Id: I39d50b8e6e55b0835670d753c3783f32b19b6c47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a050129384ac4803d1c56001a3140fc547fe134b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a050129384ac4803d1c56001a3140fc547fe134b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
