)]}'
{"id":"openstack%2Fkeystone~914212","triplet_id":"openstack%2Fkeystone~stable%2F2023.1~Icd193eff25556d21ec26bb29908b8ad6548fdc91","project":"openstack/keystone","branch":"stable/2023.1","topic":"keystone-manage-uuids-stable/2023.1","attention_set":{},"removed_from_attention_set":{"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2024-04-05 16:48:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icd193eff25556d21ec26bb29908b8ad6548fdc91","subject":"Add ability to create users and projects from keystone-manage","status":"MERGED","created":"2024-03-26 14:56:56.000000000","updated":"2024-04-05 16:50:34.000000000","submitted":"2024-04-05 16:48:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914212-keystone-manage-uuids-stable/2023.1","cherry_pick_of_change":912023,"cherry_pick_of_patch_set":4,"meta_rev_id":"ff07186c400ba615e3a36449f5c715002967a460","_number":914212,"virtual_id_number":914212,"owner":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-05 16:48:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2024-04-05 15:18:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2024-04-04 14:17:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2024-04-05 15:18:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-26 16:19:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-26 16:53:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-04 14:17:26.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-04-05 15:18:12.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"6f92433b4800169d3b12db0bddb29d0e06e2b667","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-03-26 14:56:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ae4d898c03411cf7ac0306be8f7a373f78dc4e34","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 16:19:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e228a99bd494853bd154eb4a7c1e4f1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5e18b3790a54a85baa0b87496c07db8 : TIMED_OUT in 1h 00m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7b220975bb0a47f394dc9bb389b0eec26863f99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 16:53:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04f1792cd44445db9711a145e787dddd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21d21723cf7b44b5b00d56a41020643b : SUCCESS in 16m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92dc2c5125e64298a14f16c6a6cec5f6 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8c4328f08f04f89b820fd295e50b7e3 : SUCCESS in 10m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb5a84b3ddb849a699ff687680b7e9bb : SUCCESS in 12m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/e9ac94224dc24c54b4e7a5426933ec37 : SUCCESS in 1h 02m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7baed390fd64792bc159aa59e970102 : SUCCESS in 1h 29m 38s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8000f63d230c44fbbc5c537ebc2386c4 : SUCCESS in 37m 07s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/03712b06175943dabccf4748712898b7 : SUCCESS in 39m 14s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e52b1ac821746b88436f3afa089264d : FAILURE in 26m 53s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/b54081e30d6b4b03bf49a937ea7979cc : SUCCESS in 34m 44s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/6f2809ae3e2244beb4cb31a441d5bef6 : SUCCESS in 10m 41s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/1c9e01cfe9b1439e937ea8b200f3ec01 : FAILURE in 13m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0cc30aebce42421b897ba7e7a56f25aa : SUCCESS in 57m 32s","accounts_in_message":[],"_revision_number":1},{"id":"515f92cc343d794518f27ea4be85f5f906475997","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-04-04 14:17:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ff629a3d9bcd0d6b78ff030529c60eb3f9dcb88a","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-04-05 15:18:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dd7faae5e0d943887c1dc104964f1bf33aa485cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 15:26:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a414deed645bbcf2dd6eae0ee3f77b4ad65a167","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 16:48:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc03a1f5d6e2420d91d8bbf33de3d198\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ddecdd0f28c4fd394bf05eb60c2057d : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e6ad71a12e24fe9a782271c75d2667d : SUCCESS in 11m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/270207c1929046b79b0306039eec7796 : SUCCESS in 12m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/05e7718378844cdcbdf0a6782ae2ea6f : SUCCESS in 42m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/16638e42c4eb4df2afbbff7b97e282f5 : SUCCESS in 1h 04m 35s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b4ce32a447f243c39cf3c969ae87fa6f : SUCCESS in 20m 12s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/dc89d701e59d40d18cd8712fe3257506 : SUCCESS in 28m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/faefdfe85f1e427e9886ec89ca08b4e2 : SUCCESS in 36m 25s","accounts_in_message":[],"_revision_number":1},{"id":"6966c693aa95fd4218e7fb0f0a35945c394c67a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 16:48:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ff07186c400ba615e3a36449f5c715002967a460","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 16:50:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4095c2f6eaf54fe588809f0a35563465\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e8376dea1a84db9b5761419677277e4 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b2f7aa4b13e7f756d5e5742aefa8a82f5c74287","revisions":{"9b2f7aa4b13e7f756d5e5742aefa8a82f5c74287":{"kind":"REWORK","_number":1,"created":"2024-03-26 14:56:56.000000000","uploader":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"ref":"refs/changes/12/914212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/12/914212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/914212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/914212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/914212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/12/914212/1"}}},"commit":{"parents":[{"commit":"60b964af7078278271309842670cc041832f4631","subject":"Merge \"Fix old arm64 job template\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/60b964af7078278271309842670cc041832f4631"}]}],"author":{"name":"Dave Wilde","email":"dwilde@redhat.com","date":"2024-03-07 20:27:21.000000000","tz":-360},"committer":{"name":"David Wilde","email":"dwilde@redhat.com","date":"2024-03-26 14:56:56.000000000","tz":0},"subject":"Add ability to create users and projects from keystone-manage","message":"Add ability to create users and projects from keystone-manage\n\nThis adds the ability to create users and projects directly from\nkeystone-manage.  We also add the ability to specify specific UUIDs\nfor both users and projects via the creation functions.\n\nChange-Id: Icd193eff25556d21ec26bb29908b8ad6548fdc91\n(cherry picked from commit a8366c4827b1455a9daa89d2f49f9cd1b7673685)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9b2f7aa4b13e7f756d5e5742aefa8a82f5c74287"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9b2f7aa4b13e7f756d5e5742aefa8a82f5c74287"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
