)]}'
{"id":"openstack%2Fkeystone~914214","triplet_id":"openstack%2Fkeystone~stable%2Fxena~Icd193eff25556d21ec26bb29908b8ad6548fdc91","project":"openstack/keystone","branch":"stable/xena","topic":"keystone-manage-uuids-stable/xena","attention_set":{},"removed_from_attention_set":{"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2024-03-26 18:34:32.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Icd193eff25556d21ec26bb29908b8ad6548fdc91","subject":"Add ability to create users and projects from keystone-manage","status":"ABANDONED","created":"2024-03-26 14:58:07.000000000","updated":"2024-03-26 18:34:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":912023,"cherry_pick_of_patch_set":4,"meta_rev_id":"b13d093eedec3c7f2b18ca5ba05d95717e9c8d7f","_number":914214,"virtual_id_number":914214,"owner":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-26 16:02:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-26 15:40:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-26 16:02:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-26 18:34:32.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"003a67604dc5e90093c104ee1012f8f8d7e4ebac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-03-26 14:58:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ecaba9dd4b2e1e3ea9ad93706a1a9c4536995a8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 15:40:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1ca8979610741ca90d0ecb301f9b6f3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6fa9a09f002944a3a8ef1bde408e1820 : SUCCESS in 18m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c0cebc92e2a4f06bc6fb9d0c235f7c4 : SUCCESS in 18m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16077ad06878d2ed5c56346aa36a5d1ac63f9e6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 16:02:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/32dd274210d54260928aaef0feaa54f9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55de67801fa84e499141c161241f214f : SUCCESS in 17m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/050fb387183f4eda9854fad1621316ca : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b668e4d284474f639a6a2a535644d0af : SUCCESS in 13m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/382440da6d204d6b88c13035dbdfaa0a : SUCCESS in 20m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a2b6aceda014e978ad99a02c08c9806 : SUCCESS in 14m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0c44034d3b140569caf9c44242edfba : SUCCESS in 13m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/72fc00b9e9c34f82bff72e99de61222c : FAILURE in 22m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2254b0c0c7d14f30962dd26b4e4bc0b3 : FAILURE in 10m 15s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/68087ef85bf749a992d21b72ffe2a74c : FAILURE in 9m 25s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/88e6ad44d03a4e3d9ae04e34123daead : FAILURE in 29m 31s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7ba31eaa1554f4581c71ddff6d9922c : FAILURE in 9m 18s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/be0a584958dd485f88e570ec82dafce0 : FAILURE in 10m 30s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/3526ffdfc9854648bc6f76120c98cdfb : FAILURE in 17m 30s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c70539b9b0644318a7b662e1b69b2283 : FAILURE in 20m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13fc80ac02b1460a83727d68b477b21b : FAILURE in 10m 19s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/eb5b79cf63234a90881bf70c212fe90f : FAILURE in 20m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b13d093eedec3c7f2b18ca5ba05d95717e9c8d7f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-26 18:34:32.000000000","message":"Abandoned\n\nstable/xena branch of openstack/keystone is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/xena if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bdcfb150943f302d420276f224783ce8aab18c18","revisions":{"bdcfb150943f302d420276f224783ce8aab18c18":{"kind":"REWORK","_number":1,"created":"2024-03-26 14:58:07.000000000","uploader":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"ref":"refs/changes/14/914214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/14/914214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/14/914214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/14/914214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/14/914214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/14/914214/1"}}},"commit":{"parents":[{"commit":"5f5e43b2902512288ee8aac1a0968a56bd3cbf8c","subject":"Merge \"Remove unused old job templates and experimental jobs\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5f5e43b2902512288ee8aac1a0968a56bd3cbf8c"}]}],"author":{"name":"Dave Wilde","email":"dwilde@redhat.com","date":"2024-03-07 20:27:21.000000000","tz":-360},"committer":{"name":"David Wilde","email":"dwilde@redhat.com","date":"2024-03-26 14:58:07.000000000","tz":0},"subject":"Add ability to create users and projects from keystone-manage","message":"Add ability to create users and projects from keystone-manage\n\nThis adds the ability to create users and projects directly from\nkeystone-manage.  We also add the ability to specify specific UUIDs\nfor both users and projects via the creation functions.\n\nChange-Id: Icd193eff25556d21ec26bb29908b8ad6548fdc91\n(cherry picked from commit a8366c4827b1455a9daa89d2f49f9cd1b7673685)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bdcfb150943f302d420276f224783ce8aab18c18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bdcfb150943f302d420276f224783ce8aab18c18"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
