)]}'
{"id":"openstack%2Fkeystone~914215","triplet_id":"openstack%2Fkeystone~stable%2Fwallaby~Icd193eff25556d21ec26bb29908b8ad6548fdc91","project":"openstack/keystone","branch":"stable/wallaby","topic":"keystone-manage-uuids-stable/wallaby","attention_set":{},"removed_from_attention_set":{"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2024-03-26 18:27:08.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Icd193eff25556d21ec26bb29908b8ad6548fdc91","subject":"Add ability to create users and projects from keystone-manage","status":"ABANDONED","created":"2024-03-26 14:58:21.000000000","updated":"2024-03-26 18:27:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":912023,"cherry_pick_of_patch_set":4,"meta_rev_id":"19f2a657d556ecfac41c0af5b1a6908215cb27e0","_number":914215,"virtual_id_number":914215,"owner":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-26 16:29:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-26 16:29:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-26 18:27:08.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"991086417bb4c0494f9070ba87e7ab1972872e43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-03-26 14:58:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"08ddea4e5f9448141c4bbc5ed5eec82bbf28507e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 16:29:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/44da8c1edad34e928b50594709e6d4f1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d347d43b542148159075679535d138f2 : SUCCESS in 37m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5aa3872a76fa4d4185639354c8f87cd5 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9263bf41ff9545c0bcd2965f10ba653e : SUCCESS in 25m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0db5eadcafea46bebb06a0e7ef9e5e79 : SUCCESS in 11m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f4cb2cfbbba4bc7b6d536f73a7286b3 : SUCCESS in 13m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27398db39f094a12b6704d17699ba909 : SUCCESS in 13m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/a807b7adf4394a7283f7e170dd1f34f9 : FAILURE in 28m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d0fa470213f43bf93fbb358905c22d8 : FAILURE in 17m 03s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f48be8cbcfda4fdfa4a36057117db466 : FAILURE in 22m 44s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/5554e86233754ecb9c4be7e88bd27942 : FAILURE in 28m 54s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f9e93ffb694445489c804ffedaf33bd : FAILURE in 18m 30s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/0c06e7b923d84a338f2a6b20110a9357 : FAILURE in 19m 58s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/bf8ecea7f5da4e6cabba36c3023c4ffd : FAILURE in 21m 43s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/3ac363d188a44d9cb4929dc2250694b1 : FAILURE in 9m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7beddd534ea94bde86c728b635263776 : FAILURE in 14m 19s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/5ed15d1e4b0d4210bafdfa0f749c9022 : FAILURE in 12m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6801f7d8fe6252ce650a15f14a0d50d55cb0c386","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 16:30:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38fcfda1dc66418896843436d6ea78f1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72692ad0c0cd4255bc8798a1d2b269c8 : TIMED_OUT in 1h 01m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1560cb00a1cc4dbd9c8d1cc42538eec4 : SUCCESS in 18m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"19f2a657d556ecfac41c0af5b1a6908215cb27e0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-26 18:27:08.000000000","message":"Abandoned\n\nstable/wallaby branch of openstack/keystone is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/wallaby if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"65e3596cceb885697c96f512ba7fdea0e1c91eb8","revisions":{"65e3596cceb885697c96f512ba7fdea0e1c91eb8":{"kind":"REWORK","_number":1,"created":"2024-03-26 14:58:21.000000000","uploader":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"ref":"refs/changes/15/914215/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/15/914215/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/914215/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/914215/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/914215/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/15/914215/1"}}},"commit":{"parents":[{"commit":"8c5d283af91478e1203491d656ae6b236e2d84fb","subject":"Remove unused old job templates and experimental jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8c5d283af91478e1203491d656ae6b236e2d84fb"}]}],"author":{"name":"Dave Wilde","email":"dwilde@redhat.com","date":"2024-03-07 20:27:21.000000000","tz":-360},"committer":{"name":"David Wilde","email":"dwilde@redhat.com","date":"2024-03-26 14:58:21.000000000","tz":0},"subject":"Add ability to create users and projects from keystone-manage","message":"Add ability to create users and projects from keystone-manage\n\nThis adds the ability to create users and projects directly from\nkeystone-manage.  We also add the ability to specify specific UUIDs\nfor both users and projects via the creation functions.\n\nChange-Id: Icd193eff25556d21ec26bb29908b8ad6548fdc91\n(cherry picked from commit a8366c4827b1455a9daa89d2f49f9cd1b7673685)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/65e3596cceb885697c96f512ba7fdea0e1c91eb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/65e3596cceb885697c96f512ba7fdea0e1c91eb8"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
