)]}'
{"id":"openstack%2Fkeystone~914759","triplet_id":"openstack%2Fkeystone~master~Ifc25cf32ffcdb3b8a62d6741bc38e14bca0d7763","project":"openstack/keystone","branch":"master","topic":"bug/2059780","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-17 11:23:21.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-05-03 16:38:44.000000000","reason":"\u003cGERRIT_ACCOUNT_7973\u003e replied on the change","reason_account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-04-30 18:02:18.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"Ifc25cf32ffcdb3b8a62d6741bc38e14bca0d7763","subject":"Allow domain admin to view roles","status":"MERGED","created":"2024-03-29 14:34:34.000000000","updated":"2024-05-17 11:23:21.000000000","submitted":"2024-05-08 17:04:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"914759-bug/2059780","meta_rev_id":"de1fc161e7df7b73b97931347a1c0303900f4cc7","_number":914759,"virtual_id_number":914759,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-08 17:04:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2024-05-03 16:38:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2024-05-08 15:42:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-30 18:02:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2024-05-08 15:42:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}],"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-29 14:47:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-29 15:50:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-17 15:49:07.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"CC"},{"updated":"2024-04-17 15:49:07.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2024-04-18 21:20:16.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-05-08 15:42:02.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-05-17 11:23:21.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"}],"messages":[{"id":"02c1d1b9462602d39a60e38b420681c75803a4e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-29 14:34:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76be4f8beef2269b2716a5f70f05a9b54ab687f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 14:47:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c43a222a8674931972d77b6202c45b3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b6dd5266b747411498e4e3b5970d272d : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/010c4c6d5bbe4f78814be99d697cfbc8 : SUCCESS in 11m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7593083f67e78fd657692a5a54080790cdd68026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 15:50:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2af09b16d3ac40f090524ca62702e0e5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7dbb3ae606944bba37b4162e4c9427f : SUCCESS in 34m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc5d832aa00e458788c849170ca610e9 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb0419fa20c145b3b1764b9d583ef900 : SUCCESS in 11m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36ed3e1aa6f64b1c97c86801f1b5ec4d : SUCCESS in 11m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae149ed176cf429e8f6999e300be971b : FAILURE in 12m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af33787ab0b742b381a69b8520a80400 : SUCCESS in 13m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/5efc06abcb82480487670f7417729410 : SUCCESS in 1h 03m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1190854c87d417a83998d9747368965 : SUCCESS in 1h 14m 17s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8217eb1169df4da19b191af46f095ae2 : SUCCESS in 13m 59s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/9c8a0987c4514528a2041c5061fb620d : SUCCESS in 23m 36s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/6fb01f80fb2d47898601b4d6af7f0def : SUCCESS in 33m 55s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/411e7bce0a2e4daebdaca1757fc467e0 : FAILURE in 16m 48s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/4e8aa11131f44e77a5711beebe981964 : SUCCESS in 26m 01s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e1fed37567ab444b8552285273f8d348 : SUCCESS in 11m 20s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/4d0f28c2b77844bbbce4f3912f63d98a : FAILURE in 17m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58cfab1ef5234d429625b82517fc6106 : SUCCESS in 43m 23s","accounts_in_message":[],"_revision_number":1},{"id":"baca87e8a8c3706680ec6010b54f57c36a205fca","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-17 15:25:20.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"eabb277e11a5d8c207cf2ba16f59382f31e444fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-17 18:17:51.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"62f54916b9dad782484a7f85b57311d3ef0cc7e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 18:51:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39ed7d292c1947618b8fa96cdc65a683\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e8d5cfbf9ccd4f2294cab79f847eb986 : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bc0cd5f06613467199272e14aea8d23c : SUCCESS in 31m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1ff9117d595d12d725dcafbf2e0cfa1b9028fe5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 20:07:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e511884de394a4f82499ba3abeb9405\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cb12c298c8b4c0fb9c393bcd79806b1 : SUCCESS in 15m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45f41d2ffb774f9cac6561edeb8b5250 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b5dbe76e95e4216a02acca48fbb2870 : SUCCESS in 12m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/941e3c6d7e154301b154398b92483f1c : SUCCESS in 9m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/14b4bbde867e492ca38b424201d0d408 : FAILURE in 13m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/315140820e4a4fb787dfe8fc73feced0 : SUCCESS in 10m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/a5e47c40137d4e419152c9df70a9d612 : SUCCESS in 1h 02m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/465e983b7fa64259a02f4a910d916675 : SUCCESS in 1h 38m 08s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ed02bfef23e34c6a81592882143b4928 : SUCCESS in 17m 09s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/0fd115bf14414ddf9139214bc0e359ae : SUCCESS in 34m 19s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/9d32fa2bc3594d49bbbfab89b36c1c56 : SUCCESS in 44m 56s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8035cd751b174582ba4032c181658f19 : FAILURE in 43m 51s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/18558caf43c543b481cc6cab0b412db7 : SUCCESS in 20m 36s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/68ab0d341e7746b3aba560549d22dbd7 : SUCCESS in 9m 37s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/6585ec053e4f4d18bf8560ee1f7e880c : FAILURE in 16m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0de62e8dd4ff4127a3c29cce9c82ddd6 : SUCCESS in 1h 06m 34s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/0c9f5dabf5e04680a76b4b0a51816c12 : FAILURE in 43m 07s","accounts_in_message":[],"_revision_number":2},{"id":"04d8516be74fd87c5c03a9021871a08f5ed26e26","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-04-18 21:20:16.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"71bf4dfe7539484599f7e36a13c1e5f92c8d60da","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-04-18 22:18:30.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f0c459c2a0f2201cb1a621aeea1e59400a1f75ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-23 05:40:36.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3e558a7533012f3fcae2456ab442e273ad4a31e2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-23 05:53:39.000000000","message":"Topic bug/2062045 removed","accounts_in_message":[],"_revision_number":3},{"id":"10849b16c515a515ff729bc6cb4f868332b09495","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-23 05:53:47.000000000","message":"Topic set to bug/2059780","accounts_in_message":[],"_revision_number":3},{"id":"d92118b16f4964b25e37751b5a6a01d0e4aeb0f0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-23 05:55:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"376ee397da36a6e74cd53d9f38077d3d1962a348","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 06:19:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/666a5ef8e6274e41bb947c581a9f0830\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e6d23d72fa4d41c39313304b750fcbc1 : SUCCESS in 37m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8b28b2ccccc34b6692d7eb9ba25c3520 : SUCCESS in 9m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c92fa23451170402994ae6e3444e09f0daa76696","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 06:58:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a3fe13798d143318ad1554b59ace2b6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c2205eaa2794855b47e4e27416b73f2 : SUCCESS in 15m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ff37088044c4fbdb1b9bb287b6b34e9 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c49cdfe32a843e99ed9b437ec59b331 : SUCCESS in 14m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8a214eb908644c691ec02516f34de6b : SUCCESS in 8m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d933075856ec443585383a74c51f03b8 : FAILURE in 9m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41c1bff080d64a8ba2dffaa2db01c710 : SUCCESS in 10m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/c97ebf228b4e4d9b9d0e0cca9af374b9 : SUCCESS in 1h 08m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6687f5cdd424788832c67933a394ab4 : SUCCESS in 1h 15m 53s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/95a97174f768412da3275c10e4559c4c : SUCCESS in 14m 13s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/fee69801c8ef485ab902510023b20b6e : SUCCESS in 35m 47s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/1cdbf34b47a748b6adf746a7ee264c64 : SUCCESS in 42m 23s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be7b6798b8f24270bbfa84882d456c70 : FAILURE in 26m 29s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/310eda36002f47f9bc849b8bb596cdb2 : SUCCESS in 25m 31s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/bc138f05cf634fd98e5be957e73a89a4 : SUCCESS in 10m 17s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/68c772b606fe4180b8945368877281c5 : FAILURE in 22m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7704562bee6c44f8933f338d44adc53d : SUCCESS in 55m 29s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/d02432d8ae0f46b4a627adaf4f5c92a2 : FAILURE in 45m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"898f7f535ab75ca73681a8ad694cc9ef33239ca3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-23 09:09:08.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a1d7a1c948babdc1fb3750add99d57bd6350e0f3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-04-30 18:02:18.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"267579e8d1c69d8f543114454c477793a65360b1","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-05-03 16:38:44.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"cc34f2234010b989f9fe4b7c2e37a3c89f3c8f76","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-05-08 15:42:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ec08762bccbc6f6e6963687d224087e9d118320b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 15:42:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"69caf69cfcae366b1c96680b158bfc9f6c7941c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 17:04:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf22db0088ec4deb8598f2ef690663fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94eabf50b04d4e7c90e4c680df5dcfb9 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/226ca7f4417e485f9db39385481dc185 : SUCCESS in 14m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f415e51280104125bd7e64320e2f9134 : SUCCESS in 13m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcf904f39a6448168b74070b5244692e : SUCCESS in 11m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/aaea5655aa8e4195a5147f98b9c5f63d : SUCCESS in 1h 08m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd2b3052620646feb8404d0f52d0f5e1 : SUCCESS in 1h 20m 45s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/77fa76539f024301bcb7690bd6272cb2 : SUCCESS in 13m 10s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/3c6640f22bef4aee84ca5cfd32838e52 : SUCCESS in 33m 38s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/588424c6fe5c4411af3db98e8ce3e932 : SUCCESS in 44m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04088987d671440495a18e0cd8d51c21 : SUCCESS in 55m 26s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/dd08eaf0d6b64b8cb887fa1e61c741e4 : FAILURE in 29m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e3587a20ca01eef50770d24ec8714d23985f896d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 17:04:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d6f939ca0b132ea178b4b5a30c7eb63b71383345","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 17:06:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b5870eb29f248c18731e9bcc20c8c56\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f92b2d456ec7454bbecfc59bd844584b : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3},{"id":"de1fc161e7df7b73b97931347a1c0303900f4cc7","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-05-17 11:23:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"522627de3c66113d03019122735cdfc3e0d245c8","revisions":{"094e20b8ff384be40d73daa8d79c1aa79484b0e6":{"kind":"REWORK","_number":1,"created":"2024-03-29 14:34:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/914759/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/59/914759/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/59/914759/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/59/914759/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/59/914759/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/59/914759/1"}}},"commit":{"parents":[{"commit":"4121cf6cb7c3d5e585df827d72e08296664f7326","subject":"Merge \"Add ability to create users and projects from keystone-manage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4121cf6cb7c3d5e585df827d72e08296664f7326"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-29 14:32:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-29 14:32:48.000000000","tz":540},"subject":"Allow domain admin to view roles","message":"Allow domain admin to view roles\n\nDomain admins are allowed to assign roles. So it should be allowed to\nview roles.\n\nCloses-Bug: #2059780\nChange-Id: Ifc25cf32ffcdb3b8a62d6741bc38e14bca0d7763\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/094e20b8ff384be40d73daa8d79c1aa79484b0e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/094e20b8ff384be40d73daa8d79c1aa79484b0e6"}]},"branch":"refs/heads/master"},"9162501c5f842817bde1e6d49986ad1e3068add7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-17 18:17:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/914759/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/59/914759/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/59/914759/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/59/914759/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/59/914759/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/59/914759/2"}}},"commit":{"parents":[{"commit":"7a6e1a0bdc79927e5d7fd6ad7e6dda2e04c8342c","subject":"Enable protection jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7a6e1a0bdc79927e5d7fd6ad7e6dda2e04c8342c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-29 14:32:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-17 18:15:51.000000000","tz":540},"subject":"Allow domain admin to view roles","message":"Allow domain admin to view roles\n\nDomain admins are allowed to assign roles. So it should be allowed to\nview roles.\n\nCloses-Bug: #2059780\nChange-Id: Ifc25cf32ffcdb3b8a62d6741bc38e14bca0d7763\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9162501c5f842817bde1e6d49986ad1e3068add7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9162501c5f842817bde1e6d49986ad1e3068add7"}]},"branch":"refs/heads/master"},"522627de3c66113d03019122735cdfc3e0d245c8":{"kind":"REWORK","_number":3,"created":"2024-04-23 05:40:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/914759/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/59/914759/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/59/914759/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/59/914759/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/59/914759/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/59/914759/3"}}},"commit":{"parents":[{"commit":"7a6e1a0bdc79927e5d7fd6ad7e6dda2e04c8342c","subject":"Enable protection jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7a6e1a0bdc79927e5d7fd6ad7e6dda2e04c8342c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-29 14:32:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-23 05:39:58.000000000","tz":540},"subject":"Allow domain admin to view roles","message":"Allow domain admin to view roles\n\nDomain admins are allowed to assign roles. So it should be allowed to\nview roles.\n\nNote that protection job is made non-voting until the domain admin role\ntest cases are updated.\n\nCloses-Bug: #2059780\nChange-Id: Ifc25cf32ffcdb3b8a62d6741bc38e14bca0d7763\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/522627de3c66113d03019122735cdfc3e0d245c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/522627de3c66113d03019122735cdfc3e0d245c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
