)]}'
{"id":"openstack%2Fkeystone~915470","triplet_id":"openstack%2Fkeystone~master~If7474a90aa545760d2dd4eadf1e4d7d7a7f35a06","project":"openstack/keystone","branch":"master","topic":"add-initiators","attention_set":{"13478":{"account":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"last_update":"2024-04-11 17:48:45.000000000","reason":"\u003cGERRIT_ACCOUNT_9542\u003e replied on the change","reason_account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"}}},"removed_from_attention_set":{"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2024-04-11 17:48:45.000000000","reason":"\u003cGERRIT_ACCOUNT_9542\u003e replied on the change","reason_account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"}}},"hashtags":[],"change_id":"If7474a90aa545760d2dd4eadf1e4d7d7a7f35a06","subject":"Add missing initiators in api calling providers","status":"NEW","created":"2024-04-11 09:45:42.000000000","updated":"2024-04-11 19:08:04.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7beac112700120bbb1ff3c1edecaf931a225da82","_number":915470,"virtual_id_number":915470,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-04-11 15:28:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-11 10:46:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-11 14:37:42.000000000","updated_by":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"reviewer":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"state":"CC"}],"messages":[{"id":"6b8ffb5c84585c1177a64074d6fdc277e664abb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-04-11 09:45:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5fcdd5b8f2b7ea919a572bc29b502d32b2f6f91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 10:46:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbd83a8037a44a9b8d55ee739e1f5481\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a9fb06daf74429c800a5dd41895638b : FAILURE in 14m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cbcaa777b1c4629ba368a67880d4535 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cb4b02e5b814cc594c6835eb46f216f : FAILURE in 13m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c281180f8465466e86de759ec47848dc : FAILURE in 11m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/debb715a149045d585fec04669b04108 : FAILURE in 13m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81b225b39f1f42fb8a74566b8e0fad45 : SUCCESS in 12m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/224220e478514d048a9bb4a991696f07 : SUCCESS in 58m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f8bb49b6bec427a82cc89384e04e9e4 : FAILURE in 58m 56s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/77d2c6baba4745cdb4a211e3aeea33cd : FAILURE in 11m 09s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/0702f5dd08494db1a29e21dc0e850e31 : SUCCESS in 25m 35s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/45502700538f425ca3aca6818cbc57c9 : SUCCESS in 30m 10s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89183f60637e416ca85fd0793978cd43 : FAILURE in 14m 00s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/5d96dd4cfa1141ab9feba87a8b59608a : SUCCESS in 24m 51s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/3805e7ec7bd24a478d7e834fd9e0286b : SUCCESS in 8m 30s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/d1069d058cf84cbb9ff345220b16d439 : FAILURE in 15m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11c3eddd6b2646508bde936f91bcfe0d : SUCCESS in 47m 01s","accounts_in_message":[],"_revision_number":1},{"id":"595542a07696b634b31cc0d1a84b5d9cbf37f36c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 10:48:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0059d5eb16aa496cadd66327a775b61c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cef63fa32b4548398ad068ece5564452 : TIMED_OUT in 1h 00m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6d718c266c2b47ba917034be640ca4a8 : FAILURE in 36m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"128d73b034570ea616c878624a43a535ee66d9be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-04-11 14:10:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ea08d4bfd7c1df3b77c632e94f16a21764d43e07","author":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"date":"2024-04-11 14:37:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b675f9c0acf2c6f0d75b0272a3d6357718f49c73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 14:45:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78d91e443a994a43955e6501c0cc1db2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ff6a4c622104b33b0bc9e49ee5607b1 : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/124036f550d043baa602af3c4cfdb087 : SUCCESS in 13m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ee0f27783334dbc4e69d85b27a66290b4e2d5e80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 15:28:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e00befbf4c6f490cbc793257ee52fccc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa88f24dfb264323825cd7de0562f5a2 : SUCCESS in 14m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/871e9d9d3fda464888cd69fad401f609 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4783b586ac84fc9a5a6931a13009144 : SUCCESS in 13m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cac312c4473345f1bcbee8a96d53bf37 : SUCCESS in 11m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b40a72f662b54fd5af3fee1553d97b0c : FAILURE in 10m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79adab8fba8d4f93b5c9991fcda6bd13 : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/b8d3d6c0d0ca467fab99992ba91c1cfd : SUCCESS in 50m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3268522fdaf1417391260c09d3df1b1f : SUCCESS in 1h 16m 37s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/64a0bf4c31a04bf5b409ebde0cf48dd1 : SUCCESS in 14m 18s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/831dd5dc3745412f8234794719ac7d31 : SUCCESS in 24m 45s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/5ef510baa9d84730b7bdf0bd6105fb8d : SUCCESS in 44m 49s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55fa874ce32f4ca283a65b114a483801 : FAILURE in 27m 30s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/3badee634221413d9c35d81920507585 : SUCCESS in 41m 42s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/b1250983416a450990e49a87a723b21d : SUCCESS in 21m 19s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9eadb5ed4a104e20aec846ca42030355 : FAILURE in 17m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40b3eff120314895a3b0caa5641dfb81 : SUCCESS in 1h 11m 30s","accounts_in_message":[],"_revision_number":2},{"id":"42dd49b06ae8e017c2829e8bce6d6d7479a10c7f","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-04-11 17:48:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7beac112700120bbb1ff3c1edecaf931a225da82","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-04-11 19:08:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f5dab591b7eaa7ba3ffe8a9af7e29b0028baf76c","revisions":{"fffeb4de8892c97e63bfb67659f3fdd092ba7972":{"kind":"REWORK","_number":1,"created":"2024-04-11 09:45:42.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/70/915470/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/70/915470/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/70/915470/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/70/915470/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/70/915470/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/70/915470/1"}}},"commit":{"parents":[{"commit":"8ca73f758bb613a57815fbe4ae78e3d2afa4af49","subject":"Merge \"sql: Fixup for invalid unique constraint on external_id in access_rule table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8ca73f758bb613a57815fbe4ae78e3d2afa4af49"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-04-11 09:35:34.000000000","tz":180},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-04-11 09:42:22.000000000","tz":180},"subject":"Add missing initiators in api calling providers","message":"Add missing initiators in api calling providers\n\nfollowup for Iae525ee13dec72af6a7d70db2bb59a77c682a177\n\nsome more mutating API methods were found not providing\nan audit initiator to the underlying providers\u0027 methods.\nThese include deletion of project tags, ec2 creds api\nand os-inherit api.\n\nChange-Id: If7474a90aa545760d2dd4eadf1e4d7d7a7f35a06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fffeb4de8892c97e63bfb67659f3fdd092ba7972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fffeb4de8892c97e63bfb67659f3fdd092ba7972"}]},"branch":"refs/heads/master"},"f5dab591b7eaa7ba3ffe8a9af7e29b0028baf76c":{"kind":"REWORK","_number":2,"created":"2024-04-11 14:10:07.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/70/915470/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/70/915470/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/70/915470/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/70/915470/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/70/915470/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/70/915470/2"}}},"commit":{"parents":[{"commit":"8ca73f758bb613a57815fbe4ae78e3d2afa4af49","subject":"Merge \"sql: Fixup for invalid unique constraint on external_id in access_rule table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8ca73f758bb613a57815fbe4ae78e3d2afa4af49"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-04-11 14:07:24.000000000","tz":180},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-04-11 14:07:24.000000000","tz":180},"subject":"Add missing initiators in api calling providers","message":"Add missing initiators in api calling providers\n\nfollowup for Iae525ee13dec72af6a7d70db2bb59a77c682a177\n\nsome more mutating API methods were found not providing\nan audit initiator to the underlying providers\u0027 methods.\nThese include deletion of project tags, ec2 creds api\nand os-inherit api.\n\nChange-Id: If7474a90aa545760d2dd4eadf1e4d7d7a7f35a06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f5dab591b7eaa7ba3ffe8a9af7e29b0028baf76c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f5dab591b7eaa7ba3ffe8a9af7e29b0028baf76c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
