)]}'
{"id":"openstack%2Fkeystone~916130","triplet_id":"openstack%2Fkeystone~master~I140b302d879ce1cc1f8d8de9e666cc74278a977f","project":"openstack/keystone","branch":"master","topic":"bug/2062045","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-17 10:58:04.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-05-03 16:45:00.000000000","reason":"\u003cGERRIT_ACCOUNT_7973\u003e replied on the change","reason_account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-05-08 20:28:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I140b302d879ce1cc1f8d8de9e666cc74278a977f","subject":"Allow domain users to manage credentials","status":"MERGED","created":"2024-04-17 18:17:51.000000000","updated":"2024-05-17 10:58:04.000000000","submitted":"2024-05-08 20:28:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"916130-bug/2062045","meta_rev_id":"34f8009f4580ca43aeaf557791009c4a3c77d3a0","_number":916130,"virtual_id_number":916130,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-08 20:28:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2024-05-03 16:45:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2024-05-08 17:44:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2024-05-01 17:08:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2024-05-08 17:44:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}],"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-17 18:47:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-17 19:30:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-18 01:09:37.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2024-04-18 22:18:38.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2024-05-01 17:08:03.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-05-08 17:44:57.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-05-17 10:58:04.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"}],"messages":[{"id":"85eeb73a3d666243bd0285e7e2b24eb246c693c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-17 18:17:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c71b4939ea564a531f23a0c630ab5a6438734b9c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-17 18:19:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b3ce01fbd1fc865251202ba443766f29c4f6495","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 18:47:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccee74703d4d493bb9a993c1ea8dec0c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb8328d3a7ff4120bd1c7d8135a24383 : SUCCESS in 9m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cc709a40b3f747138daaf78680990a4a : SUCCESS in 27m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04385319d9c6b5fb9b3b934d53772fd43d469a00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 19:30:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/057ba0af92fb4dad83100d3ad71d8b75\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03ab92eca50245819131f7b947061112 : SUCCESS in 16m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4874808340654a42b54fbe55cba69a52 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7efdc44465a84e07a40cacc1377f65b5 : SUCCESS in 11m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5b7d68108e5c4115bae502b1651aa107 : SUCCESS in 7m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/10bca24954144f2ea85f96f18263503c : FAILURE in 12m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d0f3b908fd40b5a0a6d213c27424ac : SUCCESS in 12m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/193b923307fe4e129167833cc74e9153 : SUCCESS in 1h 00m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b27e741a88044be58891c73b4e485364 : SUCCESS in 1h 11m 09s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b78627bcb1184ad79287bbb4469ec716 : SUCCESS in 13m 16s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2b052641aaed4c029a8fb64bb6559065 : SUCCESS in 21m 20s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/82b1bf7840834f9cb60dc17d635bcc0f : SUCCESS in 42m 55s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a61b885cc344f0e9c6b83b740bc3f70 : FAILURE in 26m 35s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/e6081386fa2047b6abea2ec3b5c0189c : SUCCESS in 22m 38s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/a42660d4275648979f47756e15a5df10 : SUCCESS in 13m 08s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/32f2c44440ad4fe79663c95d6d33a266 : FAILURE in 11m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67ddfe97ea604ea6806dd7bd589b4bfc : SUCCESS in 47m 50s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/9beac21cbc17449aa272bfcb794a1824 : FAILURE in 42m 22s","accounts_in_message":[],"_revision_number":1},{"id":"1e800cab122dab5d5cb0b733ce3de979644bfaf0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-04-18 22:18:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5dd1f68b61aebb76308f402272097da08faed2cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-23 05:45:40.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cc851934d9d4627fc84d6c7f101ca68cf2f7d96a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 05:58:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a5b58890e8d440eb9f27f397d42086b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d50168869354681b5fd1578d8cb5c27 : SUCCESS in 11m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2f40021896384b6a932ee8d3e95f9147 : SUCCESS in 8m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"81832a498f0a00783953241fa77cbcd4a2890b66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 07:21:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a21fbacc5a164c5b8b001a40f500feb1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e82e26bb5c2c47d9948f4648e0ffa646 : SUCCESS in 13m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/730bd491863442048317f8d5c5367264 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fd6093a2b914ae4a07d3be7ed7d737a : SUCCESS in 12m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/385d46b40adf493ea9ed77e17b4dce23 : SUCCESS in 8m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7639352907b741c589a7bd6bc4f03c16 : FAILURE in 12m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad85dff169524c748f13f203b0e4af24 : SUCCESS in 12m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/cecc2e0d00344809b5df7e712bb8ce18 : SUCCESS in 47m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/efe4190b47174c45a1ce2a79416a83f5 : SUCCESS in 1h 34m 16s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b2d0a2062d214ac89ce1ae656dc7c8d4 : SUCCESS in 11m 23s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/cc948497b614485bacfa28ed71e2b331 : SUCCESS in 21m 58s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/48becef3ed6e41a09d62c238a8dc88ec : SUCCESS in 31m 16s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b086312244144314867e90664c2ba36f : FAILURE in 18m 43s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/572d3fa5fbfb462f8bf5c968f6053d51 : SUCCESS in 34m 10s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/1c2bfdd254f14b3a96a8716edcc56e4b : SUCCESS in 19m 39s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9414885619f74db49db225690701839d : FAILURE in 16m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5202b6d490834f569574d1026905e0b3 : SUCCESS in 1h 03m 18s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/46dba613eaa84f11b151a656039b6e3a : FAILURE in 38m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d30a1492f47d9aeec0c632d412692f76daa77150","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-01 17:08:03.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"57a8fd12d632cd50de31d36a168a603dac67c20c","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-05-03 16:45:00.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e03bf668fd641167fc01759fe0644d6c5200acc0","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-05-08 17:44:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a53fca0b1d2dd893400a53b6b7425daf99a93b58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 17:45:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fe70116555943ce20c854dbe16ce6aa14556aa34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 20:28:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d081caf78fe8456b8084fc63be3053d5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/019763da1f1e466897c5b831067c74a6 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d7ab516c2224cb08325d6832cc09419 : SUCCESS in 17m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aabd0cc7dee6412ebbe555e980c3a46f : SUCCESS in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1f180b1c2254502a422011a545c061c : SUCCESS in 12m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/ae3754a0d12f4d94819f2a187b5ceef1 : SUCCESS in 54m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/044412459c47410bba91f9de0e2e3f67 : SUCCESS in 1h 28m 47s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9240862c4df845558ed620d72467893e : SUCCESS in 14m 23s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2719e89075f24c6bbbaca4a05f5349eb : SUCCESS in 23m 24s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/0b66284c7028440ca9e3422f46cde114 : SUCCESS in 27m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2158d3d1918a435f9471c3fb126add1a : SUCCESS in 49m 27s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/9920253961b340a6acafb44cb43d6e4c : FAILURE in 37m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dc0c74d74504f20e9e95f2e6cf73dd8dd697f10d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 20:28:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cc8be3ff8dd4586fce87add8d6d70707c94fcc2e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 20:29:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a53d7fa99d744fdcbaa3d9caedc76515\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01e4e41b4ab04692a7c576293511ef28 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2},{"id":"34f8009f4580ca43aeaf557791009c4a3c77d3a0","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-05-17 10:58:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5ead95ffcc597517feed53170e2d2f77cdd311a1","revisions":{"04a71d340fa95ccc0f883fd0bc12cecb87b9d76e":{"kind":"REWORK","_number":1,"created":"2024-04-17 18:17:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/916130/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/30/916130/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/30/916130/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/30/916130/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/30/916130/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/30/916130/1"}}},"commit":{"parents":[{"commit":"9162501c5f842817bde1e6d49986ad1e3068add7","subject":"Allow domain admin to view roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9162501c5f842817bde1e6d49986ad1e3068add7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-17 18:12:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-17 18:17:36.000000000","tz":540},"subject":"Allow domain users to manage credentials","message":"Allow domain users to manage credentials\n\nCredentials are associated with users so there is no reason we prevent\ndomain users from accessing the resources. In some services like heat\ndomain admin is used to generate keystone credentials and loosing\nthe scope check is required to continue supporting such use case.\n\nCloses-Bug: #2062045\nChange-Id: I140b302d879ce1cc1f8d8de9e666cc74278a977f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/04a71d340fa95ccc0f883fd0bc12cecb87b9d76e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/04a71d340fa95ccc0f883fd0bc12cecb87b9d76e"}]},"branch":"refs/heads/master"},"5ead95ffcc597517feed53170e2d2f77cdd311a1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-23 05:45:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/916130/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/30/916130/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/30/916130/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/30/916130/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/30/916130/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/30/916130/2"}}},"commit":{"parents":[{"commit":"522627de3c66113d03019122735cdfc3e0d245c8","subject":"Allow domain admin to view roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/522627de3c66113d03019122735cdfc3e0d245c8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-17 18:12:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-23 05:45:40.000000000","tz":0},"subject":"Allow domain users to manage credentials","message":"Allow domain users to manage credentials\n\nCredentials are associated with users so there is no reason we prevent\ndomain users from accessing the resources. In some services like heat\ndomain admin is used to generate keystone credentials and loosing\nthe scope check is required to continue supporting such use case.\n\nCloses-Bug: #2062045\nChange-Id: I140b302d879ce1cc1f8d8de9e666cc74278a977f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5ead95ffcc597517feed53170e2d2f77cdd311a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5ead95ffcc597517feed53170e2d2f77cdd311a1"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
