)]}'
{"id":"openstack%2Fkeystone~916707","triplet_id":"openstack%2Fkeystone~master~I8949b738a62d9a1c30326b3b9897b1540c5cfbd4","project":"openstack/keystone","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-11 08:08:54.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I8949b738a62d9a1c30326b3b9897b1540c5cfbd4","subject":"Make protection job voting again","status":"MERGED","created":"2024-04-23 05:53:16.000000000","updated":"2024-05-11 11:23:50.000000000","submitted":"2024-05-11 11:21:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916707-secure-rbac","meta_rev_id":"0d1fb9b6a1c15d74ae9887bcc73281487499fb70","_number":916707,"virtual_id_number":916707,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-11 11:21:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2024-05-03 16:46:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2024-05-08 17:45:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2024-05-01 17:09:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2024-05-08 17:45:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-23 06:27:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-23 06:33:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-01 17:09:01.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-05-03 16:46:50.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2024-05-08 17:45:20.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"02ad891d3f303240f3a89057e8fa4f1b69c988cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-23 05:53:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"462ab935e9a89a91b72942648f4ba91cda136608","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-23 05:54:18.000000000","message":"Topic bug/2062045 removed","accounts_in_message":[],"_revision_number":1},{"id":"0be8c044001bfa236853878a1be66b636a3bb297","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-23 05:54:22.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"55f1ab9d3d7843342577010ccc8c4e77d056cb62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 06:27:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc28a32b037547f4b90a9ac4a7503cc8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/93f717d3b4fb4c6bade21c8295d2261a : SUCCESS in 31m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/995a037dd20b4bdf906600ad39f5c6c5 : SUCCESS in 30m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"74329fd3ac41d27fbe9746c4606e34ecd044adf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 06:33:01.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/916706 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"dde52d61aa4a19487f35df36514f4bbcd30d0e5b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-23 08:42:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f7cc20d53ca2b541e82ecc35056f324443d8b43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 09:06:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1e0f4c4d59d4da595aa03a7f9339f5d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/559097db8eea4a068e17c7fb3c67c9a1 : SUCCESS in 22m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ebe7d8604f1e44c5aa9803405e179b70 : SUCCESS in 15m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c45e7512e9f3f6b09ea4e9b69514b2afeea014c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 10:23:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d241bc057544aa6bc7128b380dfcde1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65b56f32314644308e83a37012be3d8c : SUCCESS in 16m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e88cba2330b94cd09eec0e14e8d0f6f7 : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d131f7413ea4e1cb9f1f7a523ce43fd : SUCCESS in 11m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e3e7ea988c2b40aead95a318d086701a : SUCCESS in 10m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/722de3541c9646b29539e57e3cf1da9d : FAILURE in 13m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e207045c90864b10ba5a53aa6fb605fc : SUCCESS in 10m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/b5494383ba5d4a6fa94fe70f679a3818 : SUCCESS in 1h 03m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72fca7f66b204f72ac380b94df15e013 : SUCCESS in 1h 34m 26s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4ac9545fe6f84d55bf7ffcc75f72c457 : SUCCESS in 12m 34s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/7a60843abeaf4953871a2eb9e63b91be : SUCCESS in 33m 58s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/45db0477b89a43a2852266be8dd6ceec : SUCCESS in 38m 44s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e2360a00e9e4e728be11ff8bc5b57c1 : FAILURE in 24m 58s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/2fff4b9fbf10466d8b4ad6d61a8c7877 : SUCCESS in 33m 54s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/da294ca670754b079817e8e22f7c8a6e : SUCCESS in 14m 33s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/1b65ad8f793242a388aa4b69725f4fd8 : FAILURE in 11m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/955897afd41f4afba1d4f0a412bc0d41 : SUCCESS in 1h 00m 04s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/20e38f2de59b4651ac5d168daa07c1b2 : FAILURE in 48m 22s","accounts_in_message":[],"_revision_number":1},{"id":"93f97a871e27bd84e67ee4d7b5a1eb2222134621","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-23 10:34:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef72c73dd081d3e0d777fbd9bc0f1a19429aea9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 11:17:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b29081937bc047229fbb797701d2224a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec251c4f374544c9aa35c276b6fbc7d3 : SUCCESS in 41m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d84fd9ff690c4c84aa0a731a56485df9 : SUCCESS in 32m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"81ec4cca5f990f132d6b58b6eeeaaa292ce8c6a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 12:45:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8dbef7b1653458bafa70bbd558a44c6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7063cfe190440df826ad09dc18f309b : SUCCESS in 14m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52c110275053492499aecdfef17ec6ae : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3e51c0aa7b54c9f9cac688903e69e8d : SUCCESS in 12m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c2da742b16b4176b6ac61ca73c39440 : SUCCESS in 14m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fc05ba0768a44c2a8c8d1609cb28d864 : FAILURE in 13m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a83859a995dc482fb8b855241a3a9d4c : SUCCESS in 11m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/91efb4b39b4043bc97d908e485d45ae9 : SUCCESS in 46m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dded3876724b4d72ab888ee5c8b2bdb9 : SUCCESS in 1h 59m 10s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4825b7c73a6d4cde9369184a56262639 : SUCCESS in 13m 08s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/4af951804d5e4b409efa1fbc37898b4e : SUCCESS in 25m 35s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/c9e45ef81a6d49f4a15a908582a04f94 : SUCCESS in 26m 49s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/def8dde590df4db8945aa6626e5097b0 : FAILURE in 26m 46s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/aec28f6b549e49abb46e35fb59853651 : SUCCESS in 20m 38s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/5c0d2c513d384a10ae74a9650a224e58 : SUCCESS in 14m 13s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/a67f9b27b58c42f68be873dec52b4f5b : FAILURE in 16m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4ca1c90e8ca4c6faf0d05d59ab4cf06 : SUCCESS in 48m 13s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/62215fafed2a4ac095a255e2fec647d2 : SUCCESS in 36m 11s","accounts_in_message":[],"_revision_number":1},{"id":"d42f9e03649433109866394a760efe14a75d6ee7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-01 17:09:01.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8231b77c81990a50cef89990327f8279b411bacd","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-05-03 16:46:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0b2585cac31a7584fc1c5dec8db4b4eb38ba75ae","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-05-08 17:45:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"deb187aa7b49195a82bc8d281c4192e8ca4d9427","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 17:46:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9cfbb2bdbbaae8aad01cdbe28580e4a6cd420f9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 20:28:37.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/916706 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"62ebe24bf76ad08e9c0e5e8a02d835f6d31a0b28","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-11 08:08:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ffaca7b65dfbc65e4d7470741a219979f96ed9d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 09:10:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7102157eea014d488fc968d5025044e1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c908bb605f2047598e74ec29774cfd9d : SUCCESS in 59m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0efc8aae7ccf4720a05518f545c7bfda : SUCCESS in 47m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e8c8d3636aab86c9138369627778266e48dea8ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 09:39:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f07fc63bf7524737b0c09ef87e0540ec\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/810abf15c43c40038c8524a43fe6c8a2 : SUCCESS in 14m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4a63cb8f33b498399d25ed622475471 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f637d4863d5a4880b3989ae813574bc2 : SUCCESS in 12m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1cfcfaabebe455aa11d65a2bb6fc12c : SUCCESS in 10m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af17d2792d7f472a80b63555cec51a76 : FAILURE in 13m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/229d76f269da46f3974fda5eccbacb9a : SUCCESS in 11m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/a6ccbababc844c07b7be1953f610d0ed : SUCCESS in 1h 02m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e478052f3a66462091d912e37ddf9520 : SUCCESS in 1h 29m 12s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cb1094ff2cdf4b67bb0ebd8e625dad68 : SUCCESS in 20m 44s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/aac87220251e4b0cb186d2a88f857ec6 : SUCCESS in 34m 03s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/f75abf39028143679cec716c3cf1549a : SUCCESS in 31m 45s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b29f1fd359e245bc88ff78f640f58c82 : FAILURE in 21m 12s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/c30f0eb3120042cc94f7642922c8bce1 : SUCCESS in 22m 42s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/3155294faf4841b8805b3079cbfd5ff4 : SUCCESS in 11m 50s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9e80f723cc124a25ad06298b44ffb7d6 : FAILURE in 13m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1174d826a9b8494caea27fbf069886d9 : SUCCESS in 1h 09m 19s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/d3002ee8c141428b8c969f3e51b58eb4 : SUCCESS in 38m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1cbc33ea4ee001a04b9426bfb355d60594686c6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 09:40:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b024d35c8a5d6ce70fbd809f269a8e156a7e042e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 11:21:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/869f33bd8a264229a16a9e6a7ddf66f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5402ad08c27e4b34a44fb313380a52f7 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa79b03789f3431f8efd001329505be0 : SUCCESS in 12m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ad84baa28e53414f9bdc19a1442f8db4 : SUCCESS in 10m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/107e5d81701744829a2ac420a9596027 : SUCCESS in 12m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/481dc3b7aa3d44c3b276a337ef0fca0b : SUCCESS in 55m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf0580a864974f3aba270f84721ecf33 : SUCCESS in 1h 36m 05s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/851fb7783e264be2aba8ff81ec992002 : SUCCESS in 13m 11s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8e0a90e144d14435b209837a154b4833 : SUCCESS in 35m 07s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/08ba50baf4194c74a8b5ae5023f6ed72 : SUCCESS in 21m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c5e4119ece049929b46287a6b5274ec : SUCCESS in 43m 13s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/b6cf623b14514303a3f2f7f0f696c39f : SUCCESS in 30m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2093e873de960dece23b31b883c8834251751976","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 11:21:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0d1fb9b6a1c15d74ae9887bcc73281487499fb70","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 11:23:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20376d818f3f4e02987a19ddcb9c3ea6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6227b71b4ab45b28a006cdbdcfc5cb3 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af53770e6242bae084f203b2f7189cc3bac8a45f","revisions":{"af53770e6242bae084f203b2f7189cc3bac8a45f":{"kind":"REWORK","_number":1,"created":"2024-04-23 05:53:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/916707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/07/916707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/916707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/916707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/916707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/07/916707/1"}}},"commit":{"parents":[{"commit":"5ead95ffcc597517feed53170e2d2f77cdd311a1","subject":"Allow domain users to manage credentials","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5ead95ffcc597517feed53170e2d2f77cdd311a1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-23 05:52:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-23 05:52:36.000000000","tz":540},"subject":"Make protection job voting again","message":"Make protection job voting again\n\nNow keystone-tempest-plugin has been updated according to the fixed\npolicy rules.\n\nDepends-on: https://review.opendev.org/c/openstack/keystone-tempest-plugin/+/916706\nChange-Id: I8949b738a62d9a1c30326b3b9897b1540c5cfbd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/af53770e6242bae084f203b2f7189cc3bac8a45f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/af53770e6242bae084f203b2f7189cc3bac8a45f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
