)]}'
{"id":"openstack%2Fkeystone~916880","triplet_id":"openstack%2Fkeystone~master~I69d662dd3c0e70d2a614655b62dd9655c438fc0b","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"8064":{"account":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"last_update":"2025-06-13 15:37:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I69d662dd3c0e70d2a614655b62dd9655c438fc0b","subject":"Fix inconsistency in CADF initiator name field","status":"MERGED","created":"2024-04-24 10:06:53.000000000","updated":"2025-06-13 15:38:16.000000000","submitted":"2025-06-13 15:37:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916880","meta_rev_id":"6a40b433a61e672c8965002aedec66fa81ced8ab","_number":916880,"virtual_id_number":916880,"owner":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-13 15:37:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2025-06-13 14:11:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-16 15:52:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2025-06-13 14:11:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-24 10:18:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-24 11:36:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-16 15:52:29.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-06-13 14:11:28.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"6fe4d817b086f76195e21351c55eb930743ef8c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2024-04-24 10:06:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"295ef714ff96e246c0f8b31f75b204f4ea5202f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 10:18:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/587e80af8d9542cfa3cca4e02d0f5f8f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf7fe385ecae4ff38b9035c18a251d67 : SUCCESS in 10m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6cb167aed7d6437792dafbdf49aabeca : SUCCESS in 9m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c81e4b798f0de9230026e74c34f2ee9354e53854","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 11:36:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39b6184c931f4b5da44fefc7dc974666\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fe188c894c74e408b9472456a558d41 : SUCCESS in 13m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1795e19883ce4d84a57d3ebadf2717d8 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4eeb097159314abdbf373c244aab25fe : SUCCESS in 12m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5a519fcfc274674b9f77d9c6d5a167a : SUCCESS in 10m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2bc22a9e3efa4e2183ba9b1c86c51778 : FAILURE in 12m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01afce1b8924476daad0825a1140f7f6 : SUCCESS in 12m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/de91e057b3f34fa0b6a73c5c3e02e5c4 : SUCCESS in 1h 07m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3cd274d019aa4c2cbfb8a45040437ef3 : SUCCESS in 1h 09m 15s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cd66944a3ab043ec9bc3a3c29d9b4b9d : SUCCESS in 13m 07s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/c86c37e4e9784c9184f5fa63f9c2c41c : SUCCESS in 38m 08s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/defe1d25f2c245feabbe819389ca39b6 : SUCCESS in 49m 35s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1dba01445cf41158513e37dd41f4646 : FAILURE in 27m 05s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/ea40755713af4fd9877a1f18efc53698 : SUCCESS in 36m 11s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/2258dcc00ff24422b618475496446f95 : SUCCESS in 9m 33s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/1448dd56d7e9454fa9fc960e84de6e4c : FAILURE in 16m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/882c653b9d1d4aa5bded976fd33e8649 : SUCCESS in 55m 43s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/d6232078b22b4ac08142e875892c9fef : SUCCESS in 42m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4cee710244f93cc226cfbf737a4464d08aa6ce2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2024-06-12 10:33:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c5fa0b9a0d4b185b92b0187d80fb86d089a8e7e2","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2024-06-12 10:33:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a410c0c74f5e2b3b67b1903c5de97cd874f5002f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 10:47:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86ab579964804e73bc04638680da8fc5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6dc378776a454a519520d4ff9cec4e9c : SUCCESS in 9m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fe47cafef452426f8abca761579bc567 : SUCCESS in 13m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c906bc472701b4298484174f0d3a5584df9c8663","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 12:06:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a938db5952b94b72bd8216004c5fde10\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/468baf7cc86e4c5d9ccaf21460f8f758 : SUCCESS in 15m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98d6bf62340e409dbb61d8ab25a0c503 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc9561f74b3e41298d2251b7e8aba6fa : SUCCESS in 11m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c61500f8446845c7983e01e3cc027405 : SUCCESS in 10m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/13da9c47fee64bc597a3be513f6e38df : FAILURE in 11m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a0ab43e96264fe494214a312bdcfcdb : SUCCESS in 12m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/b123b265fb184f88a5d172ab468981d0 : SUCCESS in 58m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/973cbbe272ca4c2886aa17db1bb65766 : SUCCESS in 1h 31m 41s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1387b0342ede4a7983922ca90ff6b834 : SUCCESS in 15m 12s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/c2c5159f5f2d4be394883a528aa91ce8 : SUCCESS in 24m 06s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/e2c050f2cb3740e2bbe0120d1558af4f : SUCCESS in 39m 41s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec14bd6e2dd3480db80424e6422c26eb : FAILURE in 18m 43s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/aa15ca4f802346a3abcf719c76f6632d : SUCCESS in 28m 16s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/729cbde9fdcd4516b119b49ff2926cd1 : SUCCESS in 14m 25s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/4822571be6534afb8047230341491ec0 : FAILURE in 19m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48647f4d19624fa89c5bc976c2c9ea2d : SUCCESS in 52m 57s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/165099a5c39a4c9c9eb0833fe24d3a97 : SUCCESS in 36m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ac28eed190e4cc7f0153e2259cc64d6b2a8fbc39","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-04-16 15:52:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1fc55cd688fc60d8e7c50f4685a36a486ef9a4e3","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2025-06-13 14:11:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7bd5e016ba8b18d7eaa6f14203613cf606474ec2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 14:12:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"334ff7dc2ef64dd3327a9b78c5a00aaf3fedbdf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 15:37:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2055bb5eed645f2863ed7327d55c2e7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/583dfb9f7700406f96ab3c5cbf4e6579 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c8b65b1954847efbc395d4c382f9a57 : SUCCESS in 12m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a14b97acff7d477bb69bc44f606b87c1 : SUCCESS in 14m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65fdd5d855844d8a8ba9af2f07c52c51 : SUCCESS in 20m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/20140bfd6c424f058c92884dc69e5f54 : SUCCESS in 29m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/20dbdf056bfc4475a22ffad6cf13a6ae : SUCCESS in 1h 12m 00s\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/11385e454cfc4caf967c32cd166c825a : SUCCESS in 23m 04s\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/6c0d0c86f0c54f63a42411052b0be9c2 : SUCCESS in 22m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc6448a0605647bb9e54529a74b5f2c2 : SUCCESS in 58m 17s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/bd366472533e49da90da0f27cd59558a : FAILURE in 32m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"52df1d39d91ce66e1c76ae32301d57335f74a4c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 15:37:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6a40b433a61e672c8965002aedec66fa81ced8ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 15:38:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a86a1b2ec5a9475ca6ff07bbd9f70051\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0a7d1f98dc649068501adbaba59fc43 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9f35c9b2901cf75c103d90f46c8cc88f83814783","revisions":{"d731542f039a50f746bb59a09002d31e8c67b04a":{"kind":"REWORK","_number":1,"created":"2024-04-24 10:06:53.000000000","uploader":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"ref":"refs/changes/80/916880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/80/916880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/80/916880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/80/916880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/80/916880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/80/916880/1"}}},"commit":{"parents":[{"commit":"7a6e1a0bdc79927e5d7fd6ad7e6dda2e04c8342c","subject":"Enable protection jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7a6e1a0bdc79927e5d7fd6ad7e6dda2e04c8342c"}]}],"author":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2024-04-24 09:58:54.000000000","tz":600},"committer":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2024-04-24 10:06:13.000000000","tz":600},"subject":"Fix inconsistency in CADF initiator name field","message":"Fix inconsistency in CADF initiator name field\n\nFor CADF messages, keystone stores the initiator username in the\n\u0027initiator.username\u0027 field, and keystonemiddleware stores it in the\n\u0027initiator.name\u0027 field[1].\n\nCADF specs says it should be initiator:name, so make it consistent.\n\nKeep the initiator.username field so as not to break existing\ndeployments.\n\n[1] https://opendev.org/openstack/keystonemiddleware/src/branch/stable/2023.2/keystonemiddleware/audit/_api.py#L290\nCloses-Bug: #2063321\n\nChange-Id: I69d662dd3c0e70d2a614655b62dd9655c438fc0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d731542f039a50f746bb59a09002d31e8c67b04a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d731542f039a50f746bb59a09002d31e8c67b04a"}]},"branch":"refs/heads/master"},"9f35c9b2901cf75c103d90f46c8cc88f83814783":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2024-06-12 10:33:22.000000000","uploader":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"ref":"refs/changes/80/916880/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/80/916880/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/80/916880/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/80/916880/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/80/916880/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/80/916880/2"}}},"commit":{"parents":[{"commit":"dd61a91864b626314c167fcfa2db8c846d689f9e","subject":"Merge \"Remove reference to devstack-gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/dd61a91864b626314c167fcfa2db8c846d689f9e"}]}],"author":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2024-04-24 09:58:54.000000000","tz":600},"committer":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2024-06-12 10:33:18.000000000","tz":600},"subject":"Fix inconsistency in CADF initiator name field","message":"Fix inconsistency in CADF initiator name field\n\nFor CADF messages, keystone stores the initiator username in the\n\u0027initiator.username\u0027 field, and keystonemiddleware stores it in the\n\u0027initiator.name\u0027 field[1].\n\nCADF specs[2] says it should be initiator:name, so make it consistent.\n\nKeep the initiator.username field so as not to break existing\ndeployments.\n\n[1] https://opendev.org/openstack/keystonemiddleware/src/branch/stable/2023.2/keystonemiddleware/audit/_api.py#L290\n[2] https://www.dmtf.org/sites/default/files/standards/documents/DSP2038_1.1.0.pdf#page\u003d14\n\nCloses-Bug: #2063321\n\nChange-Id: I69d662dd3c0e70d2a614655b62dd9655c438fc0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9f35c9b2901cf75c103d90f46c8cc88f83814783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9f35c9b2901cf75c103d90f46c8cc88f83814783"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
