)]}'
{"id":"openstack%2Fkeystone~917923","triplet_id":"openstack%2Fkeystone~master~If7b040a39679b2707565ecd0bd8a06b7db7503c1","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-01-17 16:06:14.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-01-17 16:06:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If7b040a39679b2707565ecd0bd8a06b7db7503c1","subject":"Remove templated catalog driver","status":"MERGED","created":"2024-05-02 07:41:18.000000000","updated":"2025-01-17 16:07:46.000000000","submitted":"2025-01-17 16:06:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917923","meta_rev_id":"8f03ce3ce5fd5a888ffb15f8b649db81f8b8f536","_number":917923,"virtual_id_number":917923,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-17 16:06:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2025-01-17 14:20:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2025-01-17 14:22:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2025-01-17 14:22:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-02 07:53:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-02 09:37:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-12 11:10:29.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-01-17 14:20:36.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2025-01-17 14:22:12.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"f5a4c980024e1d14de240258125ec6615426025c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-02 07:41:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c5925df9b31b045b154ab1626f6fe24090363b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 07:53:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28ddc53ab9724826ab2f71add1bc785b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eb110b0b692f472294b3acdf5a8f9662 : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e15e6f53fc064af78a899426e782acef : SUCCESS in 9m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"927ca51825d28b332ae2bde9679e9ec202216a20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 09:37:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac0cba0f25674ccab7f703df2ffab19c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/04b233b773d849daaa4d36ed191964a1 : SUCCESS in 1m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36da092c3ec24043a06b30755ecc286d : SUCCESS in 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7dc6aaa27614fd9ad91a435dfdee5a6 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4578923668624cdbae075c2e9e49ce66 : SUCCESS in 14m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b4c5e7e92d1b48eba10b516fa529dc48 : SUCCESS in 11m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d552b507a5924d30987b06fd33a95eaa : FAILURE in 12m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9061392b52e648be99a49bec6f4dcae1 : SUCCESS in 11m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/5ce37771ca31416396f61a95931fe377 : SUCCESS in 49m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/696e3283ee184fa997772d0ea21696f4 : SUCCESS in 1h 36m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/328ea97f751d4b93a6b54f8469641f94 : SUCCESS in 7m 38s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2b1845d6c3f24829a30b3c4c490656c3 : SUCCESS in 13m 23s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b0231dcb53c3474c80bf27a0725a601a : SUCCESS in 34m 11s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/a62babc763a645c4bf3f7075177eebaf : SUCCESS in 45m 00s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a370d89240104c109253f25b0b333cb7 : FAILURE in 29m 26s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/057416de4c014943976bcde85ed67800 : SUCCESS in 23m 35s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/799920ff4f794df394369d813aa9b40e : SUCCESS in 20m 00s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/b16a0e18cdff463e97039d3ca915afda : FAILURE in 23m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18b1512db5144ceab921fd193c43965c : SUCCESS in 49m 42s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/abe0031340134e4d9fbced1a2b0b340c : SUCCESS in 45m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d77ee798ba6a05fc6b4c50a5a9596ade54d1f731","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-06-12 11:10:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8ba5a8e80bf8fa8eac79568dd5f13bdcfa8301d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-29 14:14:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"17dba96d91e539a0eeef574ee93d5cae7960eef2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 15:41:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/020672404d1d43a0a0ef11341b3f5150\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fbeeb5a40b37477da7644e517272bc49 : SUCCESS in 1m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a321f0209a24157badc774d11540d4f : SUCCESS in 17m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99649f54ca8c4fc093c727e1ef934e09 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9327285d3c2a4dac931c592231947e6f : SUCCESS in 13m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9562b1e8e5bc4ab8bab5b607ee582c4e : SUCCESS in 10m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6789c1c69f4c4819bad82b63449f89ef : FAILURE in 17m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/754055dcf8564269aa2d8a403ce60dd1 : SUCCESS in 11m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/856283177d514a5d962ba03477c11a20 : SUCCESS in 1h 11m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d9c658cc5d14ed7877689a8666b90bc : SUCCESS in 1h 23m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a2fe5dc9c384422898f7aae2d3ab1dc : SUCCESS in 13m 06s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/764f6aa684854911ab06f8fb077c5000 : SUCCESS in 19m 49s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/a7600ad748584c2e9e6d28ddcfad3250 : FAILURE in 21m 17s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b4ea30c7e7d43e39c36a8f0a7e82321 : FAILURE in 19m 51s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/371f09c7bad440279251fb26ad04a02b : SUCCESS in 32m 54s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/6703f72758f246188499630d96043e2c : FAILURE in 18m 45s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/71fc2df8c1434671a66278254cf6ad58 : FAILURE in 21m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1dfc0aa43974494bb3c576bd76bd80c : SUCCESS in 1h 06m 58s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/b4f0d7f816b74e7ea0aab1de9a059d5d : SUCCESS in 41m 02s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/3c91b4267165443e85fefa3ad297ac84 : SUCCESS in 6m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c690a1141843e219d9fe380d8b41309669d358c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 16:05:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3ade679c62d455cb29e02fb263c4010\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3f739e77faf848dbb2d6b39fea4f3e8f : TIMED_OUT in 1h 00m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/935f8a6fd8cf473aa5a1a9ab2c321f05 : SUCCESS in 50m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"36982c183403808918c54ea134e6cc91ab6f4dad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-07 10:39:58.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f5eb2181d9c398df476fd0c1d7c75bba09fc59c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 12:02:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4aea9cde7f62452e82ff952f34ba4675\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/364cd4efa84c49af8f8e7bb739e287df : SUCCESS in 1m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c1ec1a44aae40ad9ef5a4954eb64787 : SUCCESS in 12m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf605122f2d458795b3fedc0d7cadd4 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4444208636c64ece92bd84712006f68f : SUCCESS in 11m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/81e92a4bdc4c41f39fcfeb0ac4c8e81e : SUCCESS in 11m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/249d317f205e463ba6a170eef793f7ef : FAILURE in 12m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dadc08ca798f49fc9453889ffc133700 : SUCCESS in 11m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/45d62a4436f34b3bb6b6af0a801b44c0 : SUCCESS in 1h 06m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9cc5bbb85e44681addf825a03217e41 : SUCCESS in 1h 21m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7001e6529d024d2683249e89ea6aa2b5 : SUCCESS in 8m 25s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e77d736f8e7b495f8ac9f4cb06bb3337 : SUCCESS in 31m 53s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/13d8a8e9551e4f4fb582012dec1e3a08 : SUCCESS in 43m 58s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ecb4049fb2d4c809128de5297381322 : FAILURE in 24m 17s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/6e9a46e6f0234938be9fbd84ef23fd7a : SUCCESS in 27m 35s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e5775950c31244258159a64b959aa56f : SUCCESS in 20m 09s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/a722077cd6594e9cabb95bb258fabac3 : FAILURE in 21m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da57aa7cbe524c3f953994330518b6db : SUCCESS in 56m 06s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/5f7227c7b531416bb13fab22edbff34d : SUCCESS in 33m 59s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/c955ef615965486295815b254ad361ac : SUCCESS in 7m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9a3e71666094a2569a6c8e653e163276c6d95ddc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 12:02:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54b36ed7b39f4393a7c0786a64bde3d9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/80fd928ae3b5428a99bca8113495537e : TIMED_OUT in 1h 00m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0055474e6ac6482d91c1e4f34c88bcb4 : TIMED_OUT in 1h 01m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3be6a5ec6760b9178453c7c5c11b6355cd8bc296","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-06 05:05:58.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"687853f662a8516414f070e9c5625a646fbecbbd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 06:08:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8795354594394720b7adb11517f1a257\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/233885af37c04de38748b04fae39971d : SUCCESS in 56m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e181fbb109b74f95addb264b5ca0d987 : TIMED_OUT in 1h 01m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7c936179bbc049402cb2e8aae583541aa3c13472","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 06:24:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba7c253ca4464a04a4a5d0b4a2760b8d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/501997cdbbd24d3ba66215bd016f9b0c : SUCCESS in 2m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6558a59914504774adf72d313482c962 : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77bee5ed10b24a3999511f7c6e2142a8 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b759223e63c64c38abe1bb30ea933d9f : SUCCESS in 10m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c48a1aff2da1436db6fe31a8894b6aa2 : SUCCESS in 8m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/373a6d4cc48f4305bf5fe21430ef3e65 : SUCCESS in 11m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afca9c403c1a4d8e830a0272a479cf67 : SUCCESS in 11m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/0d04edf61dec402b886b254458d0e9b3 : SUCCESS in 1h 00m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8fcafd7e26aa49c9b2e36eec37f03b24 : SUCCESS in 1h 17m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32b789700cf74c83ad83ea81b83f4a07 : SUCCESS in 11m 15s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/80e93be36357431880da572344017fa0 : SUCCESS in 28m 38s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/c5bc6df77b194261aa3d0ddfc202bdca : SUCCESS in 25m 38s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96e797ad43054a9ebe51e04f986d959d : FAILURE in 14m 26s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/5b2813a1bd11495094116d5b5cba21c3 : SUCCESS in 32m 39s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/43d441e14955460dbdd5cd9e2792200f : SUCCESS in 20m 08s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c1ed36932ed04d7b83cc40698c734379 : FAILURE in 12m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/455b4ae714eb469c81879b6c71732e41 : SUCCESS in 58m 58s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/5cf432e7d9d54428a4a47135b7ef3a7e : SUCCESS in 39m 25s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/130954a990104f8d827d1ab811069cd4 : SUCCESS in 5m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d20d68168da1df2e4fd3b3e5526a3ff8fca500fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-02 13:14:21.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"9d88a109760ed0aff7b6e9bb7362e0aa134c1f26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-02 13:36:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e851d02692e4e32967b6d4dce262338\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba0d23f774364fce8ab5ca7090cf5676 : FAILURE in 12m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bfe14be4b4d144fd96ca5113cd485ef0 : FAILURE in 14m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e45c1945ec06247a7803abb2ffc21f5aceff3665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-02 14:09:11.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7545e3b2c44e4661bdfc69001a062ccd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/01c513aa639d497a93620d16569cdf6a : SUCCESS in 1m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0478d555abba4ff38a2627075b9e57a8 : FAILURE in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ba8cbec6c654ba299271d4ad4c844ae : FAILURE in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc593ad9be294e13ad36f3d3e04b1c0a : FAILURE in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cad995d957324307ad2accae07fa4fa2 : FAILURE in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e30f540ab7d14c5c9b07b92f30fe615a : FAILURE in 5m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/880fa502dba74603b721ce7f1db7fbfc : FAILURE in 48m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f8ce6e7b26343489cb297bc4cbf0c7e : FAILURE in 25m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d87d19b7ab1b484c9445dccc5c675617 : SUCCESS in 11m 54s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/cb50157e009a4601aebb781c28ae1a0c : FAILURE in 24m 35s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/525c2d6cc4494211b8917816df3c2b83 : RETRY_LIMIT in 1m 48s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d8985c600a64b97ae2a11ce8ccaaeb6 : FAILURE in 15m 15s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/4829355fa03b463ca5a6337363c6ade7 : FAILURE in 12m 31s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/a7b32f338d324468888a85fab8668276 : POST_FAILURE in 19m 38s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ce5a60eb1b3a4917a32122c4995e96ae : FAILURE in 18m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3c7d536fdf447e8a55330b1eacbf75e : FAILURE in 11m 44s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/8b2008b1654a42ca93b9e26e02f7a82f : FAILURE in 11m 30s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/614e3f9ef5db40a18462d58ffef9a15a : POST_FAILURE in 5m 18s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/1012f4c9637540508191105dc5c14eeb : FAILURE in 18m 09s","accounts_in_message":[],"_revision_number":5},{"id":"158fa6ea2676daffb56bdad89d5d0a7155a39be4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-03 01:36:10.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"602b13077632c42d3f247e4d62dd9e21262c771e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 02:39:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afd0e3951ed045edb9cca3538ed80ece\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e71d4123b12e4c0dbc8e90aac16a5196 : TIMED_OUT in 1h 00m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/401e3dbd3db8452e9f6b7e73504333c2 : TIMED_OUT in 1h 00m 59s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8eecccbe1d8472484d7545a58cf12ae3f782ac9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 03:22:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68679a7e85c94219bd865a2d566e1896\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2300baa5947a46e5ac75b26dc7dbf800 : SUCCESS in 2m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6e827e1998d4a23a5444c8bf4d4348e : SUCCESS in 12m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30b2bfb6faea4a8098575cf3d9702030 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd280d07f09e4b5cbe0f0018aba4e7fb : SUCCESS in 10m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a1b22a1d87234f0b82ab7bbab82f80e5 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caf442c58b404c86b0ee88b8751b891a : SUCCESS in 10m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/af41e90b69b94e468a4fab41c56da137 : SUCCESS in 35m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5cbeb7b27564dc38febfa4d1347c4dc : SUCCESS in 1h 44m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87cceeb93ba946e68e76e82a7d1dde78 : SUCCESS in 8m 56s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a7f20fb08a6d4880b9f7871bba5d5628 : SUCCESS in 20m 04s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/a9ac5b97f10442cbb90b72c0c3b1ec99 : SUCCESS in 26m 53s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b47ed6424a1b4c6f9d0d9cd049c6e446 : FAILURE in 17m 27s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/10e4a35071da47a996e577513ec21872 : SUCCESS in 17m 53s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/1664a1cffa0542db92caafc40ff61c1e : SUCCESS in 8m 48s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/3ddd7c8e430a47838c2137e3e0fc9cbe : FAILURE in 11m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fae8dd13ecf48a2b2a912b79235df1c : SUCCESS in 52m 15s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/43fe3bbbf9c247dab7717851b5e916ba : FAILURE in 29m 06s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/5b2b9360e9a24038bdab34b1f0b311ea : SUCCESS in 5m 14s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/d77b4659de4d478eb9dce01662d789b4 : SUCCESS in 21m 34s","accounts_in_message":[],"_revision_number":6},{"id":"281fcb847e9f4cce47f7a7f6797fd28b313cb1b4","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2025-01-17 14:20:36.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"043eb9170e1c1a13826e47ad0975ae5f495e6cf0","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-01-17 14:22:12.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"4bbf8441d663b0df1da128435bf9fc715fbae236","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 14:22:49.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b64df20036ee8b37d950fa258eced8208a50579e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 16:06:14.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/790cab0aa97641749d01b1d08c31582e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/37ca5a645fc449ccb2c0e4b81dc3bb74 : SUCCESS in 2m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c273713e53043e7ad2124b4547dc764 : SUCCESS in 7m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0cd25d49f4984e0da265862d875af3d7 : SUCCESS in 14m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/caedb722435d4797b557219c90362354 : SUCCESS in 9m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb90adf24f8b47d39e1559756bf7525a : SUCCESS in 12m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/a04107de5df541edb5835bb3de759574 : SUCCESS in 59m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a113a41249e54a0b89e8d1ebafbbed30 : SUCCESS in 1h 43m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54b9bf596c6e4975b3555a84fb1eb602 : SUCCESS in 10m 45s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1b6a5eb9d87449ffb60d23f01ab8a57b : SUCCESS in 21m 59s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/a926f9306b024970b475bbabc8cf8d5b : SUCCESS in 21m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e974f4da1ac244aea9a3e8e359fb2129 : SUCCESS in 51m 07s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/0c84873e36f14b31b6d664e403e427ac : FAILURE in 38m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"05624c8cb559176cb9e2639c0dad47a26690c717","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 16:06:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"8f03ce3ce5fd5a888ffb15f8b649db81f8b8f536","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 16:07:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f03fdb1798f54aeeb023e08882bc5f16\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2eb6521b24646df805ca0dafdef35b9 : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fe7db6467354f0b833b0415700febaa : SUCCESS in 48s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"39746b74179c996954b1e5f4ed94517e0b4af2d6","revisions":{"a62760046e72ccaa4f1b35058f3fcb374e530ee5":{"kind":"REWORK","_number":1,"created":"2024-05-02 07:41:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/917923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/917923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/917923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/917923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/917923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/917923/1"}}},"commit":{"parents":[{"commit":"7a6e1a0bdc79927e5d7fd6ad7e6dda2e04c8342c","subject":"Enable protection jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7a6e1a0bdc79927e5d7fd6ad7e6dda2e04c8342c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-02 07:33:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-02 07:35:22.000000000","tz":540},"subject":"Remove templated catalog driver","message":"Remove templated catalog driver\n\nThe driver was deprecated in 2024.1 release because of the unmaintained\ndefault template[1].\n\n[1] 307296af5e170ca6b0d44fd5ec85a39bd6b5e572\n\nChange-Id: If7b040a39679b2707565ecd0bd8a06b7db7503c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a62760046e72ccaa4f1b35058f3fcb374e530ee5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a62760046e72ccaa4f1b35058f3fcb374e530ee5"}]},"branch":"refs/heads/master"},"a5cf5ea63e65da819953c8183ef5687fda5fc45c":{"kind":"REWORK","_number":2,"created":"2024-07-29 14:14:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/917923/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/917923/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/917923/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/917923/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/917923/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/917923/2"}}},"commit":{"parents":[{"commit":"f0352c5481da6e2bfd34fc0ae25a6e433b7a8983","subject":"Merge \"Enable pyupgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f0352c5481da6e2bfd34fc0ae25a6e433b7a8983"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-02 07:33:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-29 14:14:13.000000000","tz":540},"subject":"Remove templated catalog driver","message":"Remove templated catalog driver\n\nThe driver was deprecated in 2024.1 release because of the unmaintained\ndefault template[1].\n\n[1] 307296af5e170ca6b0d44fd5ec85a39bd6b5e572\n\nChange-Id: If7b040a39679b2707565ecd0bd8a06b7db7503c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a5cf5ea63e65da819953c8183ef5687fda5fc45c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a5cf5ea63e65da819953c8183ef5687fda5fc45c"}]},"branch":"refs/heads/master"},"248a3157a4777e85a43c2c4859841ed2f889f5f0":{"kind":"REWORK","_number":3,"created":"2024-08-07 10:39:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/917923/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/917923/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/917923/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/917923/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/917923/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/917923/3"}}},"commit":{"parents":[{"commit":"f45921840c17e3b967a136de594f0c7eade82488","subject":"Merge \"Re-join the strings after re-formatting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f45921840c17e3b967a136de594f0c7eade82488"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-02 07:33:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-08-07 10:39:21.000000000","tz":540},"subject":"Remove templated catalog driver","message":"Remove templated catalog driver\n\nThe driver was deprecated in 2024.1 release because of the unmaintained\ndefault template[1].\n\n[1] 307296af5e170ca6b0d44fd5ec85a39bd6b5e572\n\nChange-Id: If7b040a39679b2707565ecd0bd8a06b7db7503c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/248a3157a4777e85a43c2c4859841ed2f889f5f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/248a3157a4777e85a43c2c4859841ed2f889f5f0"}]},"branch":"refs/heads/master"},"8c7f00ac5ad5a2e38b568907b02a4ae9804becec":{"kind":"REWORK","_number":4,"created":"2024-09-06 05:05:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/917923/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/917923/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/917923/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/917923/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/917923/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/917923/4"}}},"commit":{"parents":[{"commit":"afed96b107262e9c99aad139f3b9b28ee4b08ea2","subject":"Merge \"tox: Drop envdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/afed96b107262e9c99aad139f3b9b28ee4b08ea2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-02 07:33:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-06 05:05:43.000000000","tz":540},"subject":"Remove templated catalog driver","message":"Remove templated catalog driver\n\nThe driver was deprecated in 2024.1 release because of the unmaintained\ndefault template[1].\n\n[1] 307296af5e170ca6b0d44fd5ec85a39bd6b5e572\n\nChange-Id: If7b040a39679b2707565ecd0bd8a06b7db7503c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8c7f00ac5ad5a2e38b568907b02a4ae9804becec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8c7f00ac5ad5a2e38b568907b02a4ae9804becec"}]},"branch":"refs/heads/master"},"ead85f431e6ab3f8509c1b9cc4a5ef337cd3ff03":{"kind":"REWORK","_number":5,"created":"2025-01-02 13:14:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/917923/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/917923/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/917923/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/917923/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/917923/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/917923/5"}}},"commit":{"parents":[{"commit":"b9e1f8d0fbd9c4bc8da9d35c1289c46b9775f91e","subject":"Merge \"Enable projects and domains pagination\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b9e1f8d0fbd9c4bc8da9d35c1289c46b9775f91e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-02 07:33:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-02 13:14:07.000000000","tz":540},"subject":"Remove templated catalog driver","message":"Remove templated catalog driver\n\nThe driver was deprecated in 2024.1 release because of the unmaintained\ndefault template[1].\n\n[1] 307296af5e170ca6b0d44fd5ec85a39bd6b5e572\n\nChange-Id: If7b040a39679b2707565ecd0bd8a06b7db7503c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ead85f431e6ab3f8509c1b9cc4a5ef337cd3ff03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ead85f431e6ab3f8509c1b9cc4a5ef337cd3ff03"}]},"branch":"refs/heads/master"},"39746b74179c996954b1e5f4ed94517e0b4af2d6":{"kind":"REWORK","_number":6,"created":"2025-01-03 01:36:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/917923/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/917923/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/917923/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/917923/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/917923/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/917923/6"}}},"commit":{"parents":[{"commit":"b9e1f8d0fbd9c4bc8da9d35c1289c46b9775f91e","subject":"Merge \"Enable projects and domains pagination\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b9e1f8d0fbd9c4bc8da9d35c1289c46b9775f91e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-02 07:33:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-03 01:36:02.000000000","tz":540},"subject":"Remove templated catalog driver","message":"Remove templated catalog driver\n\nThe driver was deprecated in 2024.1 release because of the unmaintained\ndefault template[1].\n\n[1] 307296af5e170ca6b0d44fd5ec85a39bd6b5e572\n\nChange-Id: If7b040a39679b2707565ecd0bd8a06b7db7503c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/39746b74179c996954b1e5f4ed94517e0b4af2d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/39746b74179c996954b1e5f4ed94517e0b4af2d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
