)]}'
{"id":"openstack%2Fkeystone~918654","triplet_id":"openstack%2Fkeystone~stable%2F2024.1~I140b302d879ce1cc1f8d8de9e666cc74278a977f","project":"openstack/keystone","branch":"stable/2024.1","topic":"bug/2062045-stable/2024.1","attention_set":{},"removed_from_attention_set":{"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2024-05-13 18:26:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I140b302d879ce1cc1f8d8de9e666cc74278a977f","subject":"Allow domain users to manage credentials","status":"MERGED","created":"2024-05-10 15:10:52.000000000","updated":"2024-05-13 18:27:34.000000000","submitted":"2024-05-13 18:26:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918654-bug/2062045-stable/2024.1","cherry_pick_of_change":916130,"cherry_pick_of_patch_set":2,"meta_rev_id":"23cc5899bd337c658166b2ddeed131dde7ea7385","_number":918654,"virtual_id_number":918654,"owner":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-13 18:26:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2024-05-13 15:55:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2024-05-13 13:38:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2024-05-13 15:55:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-10 15:10:52.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-05-10 15:24:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-10 17:03:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-13 13:38:26.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-05-13 15:55:08.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"40bd2ea3c6419877794cfadec186b460b3780311","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-05-10 15:10:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7020dbb110c9445adeb1c125759fd5c44aacddff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 15:24:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d49bdaa7580b4203916145b1f09adad1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1da58405d885454a9a91b815cf5e9750 : SUCCESS in 10m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c75b757e023e41038485f2cc8ba2af87 : SUCCESS in 12m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"303b9acc0425d8fd8d0269219123c1e2fc389fd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 17:03:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51698e5533cb48f9abf89b0528d367bd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bf2ffaeb1634941856f34a3d7ea7c3b : SUCCESS in 14m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e73768ed9da4bc9b8743e87877f5098 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da84f33911f9456da6c3022f516e19bf : SUCCESS in 11m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d5ad028b48c42ddabfa9a3110482f72 : SUCCESS in 13m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72864df28b644a1db158946cd15dc95d : SUCCESS in 13m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/aa6983d6b9c543eaa2d14fea797773ce : SUCCESS in 51m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/03b75d697b4b465696ca30f477fb1d5f : SUCCESS in 1h 46m 32s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b063dc5da3c84eb89e58bf8de126a137 : SUCCESS in 34m 08s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/16d5c9d75ad14994b9fb25c7e6b88c1a : SUCCESS in 44m 21s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23ef94d1a4ad4056be5c073c192bd9e5 : FAILURE in 26m 50s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/e67164b4f7d743faa14d18e7064c3735 : SUCCESS in 33m 30s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d38305948f174eeda4b085ec5f46b4e9 : SUCCESS in 22m 04s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/0f4899808b8a44c895b484023bcddae1 : FAILURE in 25m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61613b51e8044f71a5447ec7a558fa27 : SUCCESS in 59m 49s","accounts_in_message":[],"_revision_number":1},{"id":"253fa27dac87e20337c12a8117578ade7ef7f6c4","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-05-13 13:38:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2c3dcf127d142860ff867ea0274037f56b5a935d","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-05-13 15:55:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ac3e2189ae044304356e021c3738f4180bf50259","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 15:55:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b777b762a0c6cbbcad0378ee7d3eccbbac56a5b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 18:26:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dfdc8ef9f614516b02207151467d1f3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e89a359a07b494ba5e0fdc20b7f274e : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc2f532eefaf47d0a8198d50abffbf09 : SUCCESS in 11m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e9340710f7594e0bbc0fcc7f14a71217 : SUCCESS in 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb5a97ff62ff4de2821e7064cc445062 : SUCCESS in 14m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/03d27ccccf6c41df80f64ac87276ed27 : SUCCESS in 1h 03m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cbf625e1a9a04b49bf76d03fe93f5a05 : SUCCESS in 1h 51m 06s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/48629f663b1f4ecea014f303037818a5 : SUCCESS in 23m 00s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/a72e69154c114d529f380fb92eafb2d7 : SUCCESS in 21m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0438f3b8ed8042b699e3f8d24055c817 : SUCCESS in 1h 04m 41s","accounts_in_message":[],"_revision_number":1},{"id":"c909763969e6a85d73bd2bba3f2cbd8d2c3d51b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 18:26:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"23cc5899bd337c658166b2ddeed131dde7ea7385","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 18:27:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12efd171103149c1a2c77719967a8b9f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7f3128be1cd43deb74182bf8aa32d7b : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a13bec662e26db7cfb0a95393993ee9f46f40b69","revisions":{"a13bec662e26db7cfb0a95393993ee9f46f40b69":{"kind":"REWORK","_number":1,"created":"2024-05-10 15:10:52.000000000","uploader":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"ref":"refs/changes/54/918654/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/54/918654/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/918654/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/918654/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/918654/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/54/918654/1"}}},"commit":{"parents":[{"commit":"286e44cd19a4eb764201f57b6ebf37a38dffce5e","subject":"Merge \"sql: Fixup for invalid unique constraint on external_id in access_rule table\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/286e44cd19a4eb764201f57b6ebf37a38dffce5e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-17 18:12:19.000000000","tz":540},"committer":{"name":"David Wilde","email":"dwilde@redhat.com","date":"2024-05-10 15:10:52.000000000","tz":0},"subject":"Allow domain users to manage credentials","message":"Allow domain users to manage credentials\n\nCredentials are associated with users so there is no reason we prevent\ndomain users from accessing the resources. In some services like heat\ndomain admin is used to generate keystone credentials and loosing\nthe scope check is required to continue supporting such use case.\n\nCloses-Bug: #2062045\nChange-Id: I140b302d879ce1cc1f8d8de9e666cc74278a977f\n(cherry picked from commit 5ead95ffcc597517feed53170e2d2f77cdd311a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a13bec662e26db7cfb0a95393993ee9f46f40b69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a13bec662e26db7cfb0a95393993ee9f46f40b69"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
