)]}'
{"id":"openstack%2Fkeystone~919517","triplet_id":"openstack%2Fkeystone~stable%2F2023.2~I140b302d879ce1cc1f8d8de9e666cc74278a977f","project":"openstack/keystone","branch":"stable/2023.2","topic":"bug/2062045-stable/2024.1-stable/2023.2","attention_set":{},"removed_from_attention_set":{"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2024-05-14 17:11:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I140b302d879ce1cc1f8d8de9e666cc74278a977f","subject":"Allow domain users to manage credentials","status":"MERGED","created":"2024-05-14 13:30:40.000000000","updated":"2024-05-14 17:13:11.000000000","submitted":"2024-05-14 17:11:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919517-bug/2062045-stable/2024.1-stable/2023.2","cherry_pick_of_change":918654,"cherry_pick_of_patch_set":1,"meta_rev_id":"be370f345948a691cb6722d1071df491ffcb1ffd","_number":919517,"virtual_id_number":919517,"owner":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-14 17:11:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2024-05-14 15:38:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2024-05-14 15:46:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2024-05-14 15:46:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-14 13:30:40.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-05-14 14:31:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-14 14:49:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-14 15:38:26.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2024-05-14 15:46:51.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"1f8bca85fa4ec868bc43229c63e37e5e552e6776","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-05-14 13:30:40.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.1.","accounts_in_message":[],"_revision_number":1},{"id":"5b265d622cd9872168a3dfc5d5f96d2466f2729c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 14:31:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdfdeac42bc04a54a63f06061b65d5f4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/88d4685230c14c419c147b3628b63f89 : SUCCESS in 58m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/507bd8ff678741849c0aa21fe1d4d107 : SUCCESS in 54m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ede4d5191ff84c43d30d2d84648ba6f31909cbe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 14:49:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f527350de2e4c4d8641ee1f9cec9b7b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f330534b5cea4ceea65e2f2f99cc1d8e : SUCCESS in 15m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/171ff2fe92b54032926dbc203169782e : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b7b1cf8ff284b618937dfff999f2e51 : SUCCESS in 12m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c436dda98ea34f80b9a349203490bb61 : SUCCESS in 13m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e862ac8c38b4a4aa22f60a44f65bac9 : SUCCESS in 9m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/891cb88dae9341a49d5035c74248d0ff : RETRY_LIMIT in 1m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adcf72bd58fd4ae0a659fc61556a6a2d : SUCCESS in 12m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/6e2df19396e04626897a2dd6c72b9de2 : SUCCESS in 44m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52bdf51c5a4845e29edd0c6a1e0cd646 : SUCCESS in 1h 14m 34s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/92425454763e41498aeba091b8a544fe : SUCCESS in 31m 35s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/85e650505a2846fa848a321161404c0e : SUCCESS in 52m 48s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71d34dc7628a426584c0680d0fe585e1 : FAILURE in 26m 15s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/7711e0fcf8f04de5a686eff60ad9ee9d : SUCCESS in 25m 40s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/018df563f62f4f388073c50620fdd4e8 : SUCCESS in 20m 13s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/eb080e62c5ba4afeb85767c3d48c9d47 : FAILURE in 23m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea168e41c84b44668f1d1555e4e9123f : SUCCESS in 1h 00m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5ade0b14eaf37e6115f6efc5e78e96ec579c0a28","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-05-14 15:38:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f45d6a575585d2eaeefa565c7f60e10bff07e2de","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-05-14 15:46:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4f9ae0eaa3b911edc2e4189f66b76e50f188d51b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 15:47:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"30c6dca27f1196e1c2e00de1d6f7dfda5a499f11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 17:11:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b50079a6b664a39ab9dc74c32e6a817\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7db05703af5545eeb78a15eb0278b53f : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de4b0029a4de4c449ec1027889a3ee34 : SUCCESS in 11m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d4e5442ecc84cd18416fc72940cb061 : SUCCESS in 17m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1bfec9410e04bfd8f2d6953b53a7149 : SUCCESS in 10m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6a6b8b7e57d47b885391d8cb84a7644 : SUCCESS in 13m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/f4b5d73c0f5c46a99281168f41edd00c : SUCCESS in 55m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/37e322bc613a416aae4910d7b227bb1a : SUCCESS in 1h 24m 26s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/04a3a4a414284639a5d6884341446ccc : SUCCESS in 32m 50s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/76dc51ad10134984933dd01a0ae93b03 : SUCCESS in 32m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0a36bfd8dbc47b3bb0f3830e84af48e : SUCCESS in 1h 00m 11s","accounts_in_message":[],"_revision_number":1},{"id":"70787df808a2948cff87efac3bceaef4e2631e6e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 17:11:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"be370f345948a691cb6722d1071df491ffcb1ffd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 17:13:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/177d48dcc4c242f9b03027801282dbc6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ba1234d215841d6917028ecea791771 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4711b36ab52effbdb7f62373a1ec9299270b08d5","revisions":{"4711b36ab52effbdb7f62373a1ec9299270b08d5":{"kind":"REWORK","_number":1,"created":"2024-05-14 13:30:40.000000000","uploader":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"ref":"refs/changes/17/919517/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/17/919517/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/919517/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/919517/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/919517/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/17/919517/1"}}},"commit":{"parents":[{"commit":"ab9cf925e8031ec29393fdd07646202e87c022a0","subject":"Merge \"Add ability to create users and projects from keystone-manage\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ab9cf925e8031ec29393fdd07646202e87c022a0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-17 18:12:19.000000000","tz":540},"committer":{"name":"David Wilde","email":"dwilde@redhat.com","date":"2024-05-14 13:30:40.000000000","tz":0},"subject":"Allow domain users to manage credentials","message":"Allow domain users to manage credentials\n\nCredentials are associated with users so there is no reason we prevent\ndomain users from accessing the resources. In some services like heat\ndomain admin is used to generate keystone credentials and loosing\nthe scope check is required to continue supporting such use case.\n\nCloses-Bug: #2062045\nChange-Id: I140b302d879ce1cc1f8d8de9e666cc74278a977f\n(cherry picked from commit 5ead95ffcc597517feed53170e2d2f77cdd311a1)\n(cherry picked from commit a13bec662e26db7cfb0a95393993ee9f46f40b69)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4711b36ab52effbdb7f62373a1ec9299270b08d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4711b36ab52effbdb7f62373a1ec9299270b08d5"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
