)]}'
{"id":"openstack%2Fkeystone~919647","triplet_id":"openstack%2Fkeystone~stable%2F2023.1~I140b302d879ce1cc1f8d8de9e666cc74278a977f","project":"openstack/keystone","branch":"stable/2023.1","topic":"bug/2062045-stable/2024.1-stable/2023.2-stable/2023.1","attention_set":{},"removed_from_attention_set":{"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2024-05-16 15:52:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I140b302d879ce1cc1f8d8de9e666cc74278a977f","subject":"Allow domain users to manage credentials","status":"MERGED","created":"2024-05-15 12:06:58.000000000","updated":"2024-05-16 15:54:23.000000000","submitted":"2024-05-16 15:52:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919647-bug/2062045-stable/2024.1-stable/2023.2-stable/2023.1","cherry_pick_of_change":919517,"cherry_pick_of_patch_set":1,"meta_rev_id":"819ac11cb4453fc92459d0e449594b8cb5ce3d28","_number":919647,"virtual_id_number":919647,"owner":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-16 15:52:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2024-05-15 19:34:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"date":"2024-05-16 13:06:26.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2024-05-16 13:14:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"date":"2024-05-16 13:06:26.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2024-05-16 13:14:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-15 12:06:58.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-05-15 12:36:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-15 13:30:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-15 19:34:58.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2024-05-16 13:14:47.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"d925473f9064f2586bcf54758f730490da4106a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-05-15 12:06:58.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"3ae71d36a4744f76649e8ac9f819f8270ea09d76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 12:36:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/616972b2cf264017a1552fd3168b8aaf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c03456d938bd44b8bc0c856c1916a918 : SUCCESS in 27m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"89f2350eca8aedc86aa6c329ac2d4884a51a40a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 13:30:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31ba3a1afbbc4230aa6b98f05b3044da\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b784774c54304d02a5ccea35fa247bf2 : SUCCESS in 15m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea7bb49e0908400d8a08d90889865d4c : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/641fa783032e45b49ecb1b7d5a27aa96 : SUCCESS in 11m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac21b118f01c40d69cb9d0692db6f4ab : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/b0c3962beab44f0697de0f695cbf5af7 : SUCCESS in 1h 02m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/564eba24f28a4caea59723d91dc9efb5 : SUCCESS in 1h 18m 13s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/3d68b9674661403c93d092b0de1c2d8e : SUCCESS in 34m 03s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/bc757115f20441c5841b50c56c44ada5 : SUCCESS in 46m 43s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45c4c3f63aae407299ed2eeac8239e4b : FAILURE in 28m 05s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/891d153c35f44d75a6818264d0075f2c : SUCCESS in 30m 16s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/5e5172f61dd14e20af18ac4b7da01a05 : SUCCESS in 13m 53s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/22bd194cad454050a34d6bf58d76e98b : FAILURE in 16m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/134441424d41449cb694e85b1dcfa9ee : SUCCESS in 47m 58s","accounts_in_message":[],"_revision_number":1},{"id":"168c50c12e08fbee99a460b638bc77b83d00ea71","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-05-15 19:34:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e95f864a5f15b07238e64b585de155f19c85385a","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-05-16 13:05:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cac70ef0aacccb9f136e748edcdb3bfc34518923","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-05-16 13:06:26.000000000","message":"Patch Set 1: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"68c162be5101a506aa267f38c98961ea45cebda5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 13:06:37.000000000","message":"Patch Set 1: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":1},{"id":"94771b1828ea4b425ddd7fc19b52531199cceba0","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-05-16 13:14:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4ac287002f27e60ab584c25720f2298ff17b44c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 14:37:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e50cfa42a2b8455c9e816323a33300bc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4d37a0e30ae44149e6ad6a11b0de1aa : SUCCESS in 16m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe36120641d84355bfd85c09c7e54aa0 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12c3d8f99c8d458089ec68eb18baef53 : SUCCESS in 14m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f64de80c54ce4a15b093a563a53fe0c8 : SUCCESS in 11m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/84bfeecf02ca4619ab64eae1e13e2065 : SUCCESS in 55m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4168ed59d104ef59540ec2bfa399f2e : SUCCESS in 1h 17m 14s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ce09de2b94ac4b64b6e111a1e3d9d839 : SUCCESS in 33m 16s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/5f06687ed81542a28aba4be3fad0c560 : SUCCESS in 29m 17s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ed1ed786ca6499ba4058fe47d98e748 : FAILURE in 24m 14s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/c4434c310f7b4eb1bf0442c73e94ae7a : SUCCESS in 50m 59s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e47d4effd79b47dc8034bd89e06b8906 : SUCCESS in 22m 49s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/5ecc248a3de64f42ae42d81be67daf5b : FAILURE in 12m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80cf0d0030ad4edb80b1109e54d194aa : SUCCESS in 52m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b680b65f54e58fa19ed4c57e2b4edfd42638e9b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 14:38:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d162b62abdbe58be2b5d3cc42043da0177a62947","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 15:52:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09bac60c6226446a9af0a69b6304f8f6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df0664a1530f42b4b3cf00be156e135f : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e89a77b8425744be8db38e1bb282035b : SUCCESS in 11m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef288175ee5347ed8c304375f333e28b : SUCCESS in 12m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/9ae14702bdbe42529ddbdbfaf05cbd1e : SUCCESS in 1h 06m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a00d1b77f61b43b5b678754a630b2e77 : SUCCESS in 58m 43s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/90234c7b1a4149eeb60aaf7cf884ffaf : SUCCESS in 33m 33s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/3b9d2405ae804552810513723edfd655 : SUCCESS in 25m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d568173149c54fbab30260400abec1a0 : SUCCESS in 53m 17s","accounts_in_message":[],"_revision_number":1},{"id":"0abfedac60a53fa91aaab8edecc37b2f798b510e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 15:52:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"819ac11cb4453fc92459d0e449594b8cb5ce3d28","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 15:54:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c762203c6b5d41248bc06a500f636b36\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b70d316accc4422480062d62f285ec51 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"17da229f0d05bf0298ab5661d695015e63d7b10c","revisions":{"17da229f0d05bf0298ab5661d695015e63d7b10c":{"kind":"REWORK","_number":1,"created":"2024-05-15 12:06:58.000000000","uploader":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"ref":"refs/changes/47/919647/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/47/919647/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/919647/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/919647/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/919647/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/47/919647/1"}}},"commit":{"parents":[{"commit":"46f66b8ac420bf2a59af2d9de41701e6162c7361","subject":"Merge \"Allow admin to access tokens and credentials\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/46f66b8ac420bf2a59af2d9de41701e6162c7361"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-17 18:12:19.000000000","tz":540},"committer":{"name":"David Wilde","email":"dwilde@redhat.com","date":"2024-05-15 12:06:58.000000000","tz":0},"subject":"Allow domain users to manage credentials","message":"Allow domain users to manage credentials\n\nCredentials are associated with users so there is no reason we prevent\ndomain users from accessing the resources. In some services like heat\ndomain admin is used to generate keystone credentials and loosing\nthe scope check is required to continue supporting such use case.\n\nCloses-Bug: #2062045\nChange-Id: I140b302d879ce1cc1f8d8de9e666cc74278a977f\n(cherry picked from commit 5ead95ffcc597517feed53170e2d2f77cdd311a1)\n(cherry picked from commit a13bec662e26db7cfb0a95393993ee9f46f40b69)\n(cherry picked from commit 4711b36ab52effbdb7f62373a1ec9299270b08d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/17da229f0d05bf0298ab5661d695015e63d7b10c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/17da229f0d05bf0298ab5661d695015e63d7b10c"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
