)]}'
{"id":"openstack%2Fkeystone~919717","triplet_id":"openstack%2Fkeystone~master~I14007ad2356fd41198ac412895c6ae88232b2681","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-31 03:16:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I14007ad2356fd41198ac412895c6ae88232b2681","subject":"Remove reference to devstack-gate","status":"MERGED","created":"2024-05-15 11:02:04.000000000","updated":"2024-05-31 03:19:26.000000000","submitted":"2024-05-31 03:16:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919717","meta_rev_id":"ea0e242a049a9d03660f8eedc0db7ecc58d0ba71","_number":919717,"virtual_id_number":919717,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-31 03:16:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2024-05-31 01:27:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2024-05-29 12:17:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2024-05-29 14:36:22.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2024-05-31 01:27:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-15 11:27:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-15 12:32:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-29 12:17:09.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-05-29 14:36:22.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2024-05-31 01:27:54.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"532a5da9207c16f7ba1d22f7726b393ef807eae0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-15 11:02:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8451feb8b5099fca9d880c2536298d16a1afb7ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 11:27:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bf221076bb34a99a5014bb1cbe1af99\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5540c1824bbb4df98df1dff8a061666d : SUCCESS in 23m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1366df87ee0e43dca74c2aceae9daa96 : SUCCESS in 10m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a34807ce41a65dde4e804f145d0c217a7c69c609","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 12:32:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2241f696c85d4ea0aca52cad5740eeb5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecfefb97249343cf8751b2359b1396db : SUCCESS in 12m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a2b03de5d70448594e383e1f74fbbac : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a3ff72a30e94ae19598c90d5ab907ef : SUCCESS in 10m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ccd28b60530542ae8e32b859dcc3a2bc : SUCCESS in 11m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/df8d7b3709ac49cea5aa026245304c3a : FAILURE in 15m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed13ec4954e64f23b324ca1c36f62de5 : SUCCESS in 12m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/24551e60d9624cdb8f74ec87691edfe6 : SUCCESS in 47m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/161fdab8c20245789755f5e0fa03b8ad : SUCCESS in 1h 26m 17s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b0d63d1c59fb48d5bfe516a112d041cf : SUCCESS in 11m 41s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b49dd353c08747308f07984f01183f69 : SUCCESS in 26m 29s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/a9a2e809efbc470280c07e576b9a4e5d : SUCCESS in 30m 50s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b4cb806715643af8ce173d9dc313fa8 : FAILURE in 15m 55s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/dff2aa58feed47a596c6512216975396 : SUCCESS in 21m 16s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/fb9b2d736f6e426cb27779c8b63a5932 : SUCCESS in 10m 09s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c7c996d8310640a492602f3037268b43 : FAILURE in 11m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b68665af020843ac818fffbd22690683 : SUCCESS in 41m 20s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/8e5a32a4da02442c8095af78b3067d33 : SUCCESS in 42m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b72d90d202acbeec088bb2f0d10595491edc885","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-05-29 12:17:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9405b8612c980be737203e7923d5e0890997204e","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-05-29 14:36:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"db64abbd1e0511d70ef8fbbba5e1bce5f000f8d8","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-05-31 01:27:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0b24ab5c2bf650caab6442d677d8d30340bdeee8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 01:28:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"588b1e680b5fba2239e9d6a1e8cbba3b2674368b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 03:16:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3b65911a1d14156ae89dcaf86c6cd74\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d45e64eafdba48dbbc3e659d327cf064 : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1036eb3d3694a15b7d34f321076d224 : SUCCESS in 13m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7989e93ad3684857aba36a3eaf71071d : SUCCESS in 11m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30c3bbe515be480c811d5633cb41042d : SUCCESS in 12m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/24acbeff9fb848899a276bc6e049f648 : SUCCESS in 1h 00m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa0781e65f684fe5b6e2cdbae6e7a7a8 : SUCCESS in 1h 42m 54s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8e04711beeb6448aa9dc9b3013e4020d : SUCCESS in 11m 02s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/329eb80fc2c94a0db5c4a8c9e99e4020 : SUCCESS in 33m 45s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/5814ed4cf71b4b31b548822c5c570fed : SUCCESS in 19m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dece3f7451b445fca34bca7693cc4df7 : SUCCESS in 1h 06m 05s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/54c439938b314cd1bf53c433c2538615 : SUCCESS in 24m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"043df9faa5da86bf9c6e864b56fcb2bebdae1af6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 03:16:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ea0e242a049a9d03660f8eedc0db7ecc58d0ba71","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 03:19:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/719af6b128a74ea18d74f5dc19e5a7dd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/848439906f3243d98981835bdc0e9f33 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e00821762f236cea42caf23415de064cc6f663e","revisions":{"0e00821762f236cea42caf23415de064cc6f663e":{"kind":"REWORK","_number":1,"created":"2024-05-15 11:02:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/919717/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/17/919717/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/919717/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/919717/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/919717/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/17/919717/1"}}},"commit":{"parents":[{"commit":"7a6e1a0bdc79927e5d7fd6ad7e6dda2e04c8342c","subject":"Enable protection jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7a6e1a0bdc79927e5d7fd6ad7e6dda2e04c8342c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-15 11:01:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-15 11:01:46.000000000","tz":540},"subject":"Remove reference to devstack-gate","message":"Remove reference to devstack-gate\n\ndevstack-gate was deprecated in xena and is being retired now[1].\n\n[1] https://review.opendev.org/c/openstack/governance/+/919629\n\nChange-Id: I14007ad2356fd41198ac412895c6ae88232b2681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0e00821762f236cea42caf23415de064cc6f663e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0e00821762f236cea42caf23415de064cc6f663e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
