)]}'
{"id":"openstack%2Fkeystone~920068","triplet_id":"openstack%2Fkeystone~master~I69755adcc4a93fcc0c74dae9941ad3d0ec30f097","project":"openstack/keystone","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-31 07:39:37.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I69755adcc4a93fcc0c74dae9941ad3d0ec30f097","subject":"Replace renamed attribute_mapped_collection","status":"ABANDONED","created":"2024-05-21 02:34:57.000000000","updated":"2024-08-27 11:38:21.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a4db5a2feafe4c5b243b039a7350cda07ef5c4ef","_number":920068,"virtual_id_number":920068,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-21 04:29:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2024-05-21 15:56:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":-1,"date":"2024-05-31 13:11:13.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-21 03:25:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-21 04:29:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-21 15:56:55.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"e3b7dbfe8c95ecdeae57e38184f3842df1368e41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-21 02:34:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a08a8aa25721b0db6a50ed999a6910a83a22305","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-21 02:35:21.000000000","message":"Topic set to sqlalchemy-20","accounts_in_message":[],"_revision_number":1},{"id":"804dbbd9691bfd2ac9f6e5af7e925773966d0362","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-21 02:37:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"74f4f87a35badcc559f00afbad5d5909e4418bd6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 03:25:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e1c601cc637476297d818dfdb9f3f67\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9315c0fe2b664fff82d95022d2bc700f : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0a2c191c76ad4d53afaa3bdd221c1803 : SUCCESS in 46m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ea609d9a5219d45eb5141790c5bbfeac1d33b0d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 04:29:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51565afd82ac4d14946c581802a054a7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7ac75bb2c7e74de6920d55331661d6fd : SUCCESS in 6m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75911292c6f54de2a014a1ddcf8ab7ee : SUCCESS in 18m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34d8b69a5587400e981c7a7e8f847cd7 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b7532ac28ff448190b96017e778269b : SUCCESS in 13m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4339bb6920649b4bcaa3542ce4db27d : SUCCESS in 15m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/11820394c8a84999b6ed5ec28d1536e8 : FAILURE in 15m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc764ac71b3a45eba3c1804d73ec0c5a : SUCCESS in 11m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/831e120b0d054a51b540aeb277ff20db : SUCCESS in 4m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/7f449ee0c9094c918e8e724fc8b11641 : SUCCESS in 45m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/65327348604643a38591333fad848f15 : SUCCESS in 1h 39m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f8e31f097064888b6188af67ebc580a : SUCCESS in 10m 23s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/788d3bfdcf2842b0a0d9415da101ec9b : SUCCESS in 17m 58s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/94ebff24be6d4858a557fff2392c6c26 : SUCCESS in 19m 24s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/7ccf63fe22e54dcbad8ac1eb37a4be72 : SUCCESS in 28m 44s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2367e949c7a14e68ba7774b774d664f7 : FAILURE in 27m 22s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/99939c45eaaf4380b48c3bf3b6cccd7a : SUCCESS in 32m 41s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/55fadf9fc4e74044b93d2b32e95d7dd6 : SUCCESS in 21m 02s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ed8eac3224994f2b8e35d2c18a40b5d1 : FAILURE in 28m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8db49f8f5fea46c0813beca2bd8200f4 : SUCCESS in 52m 38s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/8daa7f9b5b124ac2ba5ef0481a73a403 : SUCCESS in 38m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0b8b2cd2cb780b3d1b8fa3d600d15fdd3aa217d8","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-05-21 15:56:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b121144faa1c63f219c36a95c4886c92b538d40f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-31 07:39:37.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"59ca158b900df4f25d33587911278350f1c8e4e0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-31 12:01:03.000000000","message":"Patch Set 2: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f0687cfaf42bf2aba665c0a7389a9b3f7161ea69","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-31 13:11:13.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"a4db5a2feafe4c5b243b039a7350cda07ef5c4ef","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-27 11:38:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6d88b50fcabaa3f91082ce3ce5cf138e41819403","revisions":{"736d4c8bd43171f0fa85751119e618f494a1847b":{"kind":"REWORK","_number":1,"created":"2024-05-21 02:34:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/920068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/68/920068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/68/920068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/68/920068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/68/920068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/68/920068/1"}}},"commit":{"parents":[{"commit":"af53770e6242bae084f203b2f7189cc3bac8a45f","subject":"Make protection job voting again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/af53770e6242bae084f203b2f7189cc3bac8a45f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-21 02:34:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-21 02:34:20.000000000","tz":540},"subject":"Replace renamed attribute_mapped_collection","message":"Replace renamed attribute_mapped_collection\n\nattribute_mapped_collection was renamed to attribute_keyed_dict in\nSQLAlchemy 2.0[1].\n\n[1] https://docs.sqlalchemy.org/en/20/orm/collection_api.html#sqlalchemy.orm.attribute_mapped_collection\n\nChange-Id: I69755adcc4a93fcc0c74dae9941ad3d0ec30f097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/736d4c8bd43171f0fa85751119e618f494a1847b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/736d4c8bd43171f0fa85751119e618f494a1847b"}]},"branch":"refs/heads/master"},"6d88b50fcabaa3f91082ce3ce5cf138e41819403":{"kind":"REWORK","_number":2,"created":"2024-05-21 02:37:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/920068/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/68/920068/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/68/920068/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/68/920068/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/68/920068/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/68/920068/2"}}},"commit":{"parents":[{"commit":"af53770e6242bae084f203b2f7189cc3bac8a45f","subject":"Make protection job voting again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/af53770e6242bae084f203b2f7189cc3bac8a45f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-21 02:34:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-21 02:37:21.000000000","tz":540},"subject":"Replace renamed attribute_mapped_collection","message":"Replace renamed attribute_mapped_collection\n\nattribute_mapped_collection was renamed to attribute_keyed_dict in\nSQLAlchemy 2.0[1].\n\n[1] https://docs.sqlalchemy.org/en/20/orm/collection_api.html#sqlalchemy.orm.attribute_mapped_collection\n\nChange-Id: I69755adcc4a93fcc0c74dae9941ad3d0ec30f097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6d88b50fcabaa3f91082ce3ce5cf138e41819403"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6d88b50fcabaa3f91082ce3ce5cf138e41819403"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
