)]}'
{"id":"openstack%2Fkeystone~922291","triplet_id":"openstack%2Fkeystone~stable%2F2024.1~I14007ad2356fd41198ac412895c6ae88232b2681","project":"openstack/keystone","branch":"stable/2024.1","topic":"retire-devstack-gate","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-06-19 09:15:00.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"35229":{"account":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"last_update":"2024-06-21 16:26:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I14007ad2356fd41198ac412895c6ae88232b2681","subject":"Remove reference to devstack-gate","status":"MERGED","created":"2024-06-19 08:50:32.000000000","updated":"2024-06-21 16:28:35.000000000","submitted":"2024-06-21 16:26:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"922291-retire-devstack-gate","meta_rev_id":"2193eb7cdeee5d870ef234aaf7df7e64566ad61f","_number":922291,"virtual_id_number":922291,"owner":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-21 16:26:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-06-19 09:15:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2024-06-21 15:00:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2024-06-21 15:00:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-19 08:50:32.000000000","updated_by":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-06-19 09:01:46.000000000","updated_by":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-06-19 09:06:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-19 10:39:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-21 15:00:55.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"973b17dc57daa7e6ddb109b7a7c6d3c4b5e371b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2024-06-19 08:50:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08c7e18983c70cc3b38eac6ccb8ba3453b73fdda","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2024-06-19 08:51:57.000000000","message":"Topic set to retire-devstack-gate","accounts_in_message":[],"_revision_number":1},{"id":"0f4fd4abd30690bd1abdd5e2db35eb3eb66e7568","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 09:06:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1d4b362bcca47919fe127a54598592e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b593f7a910d428bab99d5fec33b65e4 : SUCCESS in 12m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ef3be275c98a451ba638d13b71c73923 : SUCCESS in 14m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0f9d9f247958e0d77806515725d5b71169c6ec5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-06-19 09:15:00.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"775d7d390cb2724eb002be41e608fe1457dc6413","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 10:39:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe78a61bf0db43bb94cc7536166a95e8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f61083c161014030937efd394da8c86f : SUCCESS in 12m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ad4c8621cbd4ecc9735844ce14ae999 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc96fde89e984261ad2d325619703f3e : SUCCESS in 12m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e304b1fd621141b7a95d4dd494a5ae45 : SUCCESS in 10m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ac629a52f414600943b516545a17736 : SUCCESS in 11m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/48eea7c09e184ab0b3a7ac108ffb2a86 : SUCCESS in 52m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/340c19a51bb546f1ac60890b1df75e27 : SUCCESS in 1h 46m 34s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/71a287562a82464db27a7206e76773fa : SUCCESS in 23m 40s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/9323bb50b9c2496a8c3ebb8a3c94bbc6 : SUCCESS in 45m 57s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e104be8c6614401d8df5363e93fd3881 : FAILURE in 28m 32s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/0df791a51ae7461996502db4ade0562e : SUCCESS in 33m 09s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c49d18f5643f45aa833970e9acb28408 : SUCCESS in 11m 14s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/3cee68a3b961492c822e7e57fde7991a : FAILURE in 16m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d03c35c43c63488da6d5afa69c9cc196 : SUCCESS in 1h 00m 06s","accounts_in_message":[],"_revision_number":1},{"id":"589d5d3b92f8db637bfb76a822b33afe7a24d764","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-06-21 15:00:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f2ea709fb5fe0392e157335576cad2d50b953457","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-21 15:01:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10f19ebca0d19f6d2cf941f5815c00f6f7fdf568","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-21 16:26:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/470d2f00931e4b0493b0fe040edfa265\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc2a8b9b196741cca96dbb88bdbce825 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e552aeb5a74474493181d131d74f8ef : SUCCESS in 11m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f76eabf672a64c249876c51299478e37 : SUCCESS in 12m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9efa805f1a9441cb8fd9eb38b3e753f : SUCCESS in 12m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/aa3d5fd007f14b43ac86e0b933d15191 : SUCCESS in 1h 06m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d48a777d34634c54af39f92f2fa1716c : SUCCESS in 1h 23m 18s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1462fb7e64d947ac87f415326c429ecb : SUCCESS in 34m 59s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/d5bba36f2f954c18a721fe2270aa9815 : SUCCESS in 34m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87345b55f082400a8f3c95619fc92f1a : SUCCESS in 58m 08s","accounts_in_message":[],"_revision_number":1},{"id":"f5a25995d51a53d285d696fb5f9327c8aef0deba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-21 16:26:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2193eb7cdeee5d870ef234aaf7df7e64566ad61f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-21 16:28:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e980bd6835b2485bb2aa611bf1af83a9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/905699f560b84cec82745a94a199e21a : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f8f451455fb11fb177148cfb88bed29caf18d38a","revisions":{"f8f451455fb11fb177148cfb88bed29caf18d38a":{"kind":"REWORK","_number":1,"created":"2024-06-19 08:50:32.000000000","uploader":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"ref":"refs/changes/91/922291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/91/922291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/922291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/922291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/922291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/91/922291/1"}}},"commit":{"parents":[{"commit":"2c8fd6cec862830725c98e00009ede00b85e18e7","subject":"Merge \"Allow domain admin to view roles\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2c8fd6cec862830725c98e00009ede00b85e18e7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-15 11:01:46.000000000","tz":540},"committer":{"name":"Jakub Skunda","email":"jskunda@redhat.com","date":"2024-06-19 08:49:44.000000000","tz":120},"subject":"Remove reference to devstack-gate","message":"Remove reference to devstack-gate\n\ndevstack-gate was deprecated in xena and is being retired now[1].\n\n[1] https://review.opendev.org/c/openstack/governance/+/919629\n\nChange-Id: I14007ad2356fd41198ac412895c6ae88232b2681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f8f451455fb11fb177148cfb88bed29caf18d38a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f8f451455fb11fb177148cfb88bed29caf18d38a"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
