)]}'
{"id":"openstack%2Fkeystone~922851","triplet_id":"openstack%2Fkeystone~master~Iaf418d69cbb55fbe9dde3f2ec4213e64345057c2","project":"openstack/keystone","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-07-26 16:25:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaf418d69cbb55fbe9dde3f2ec4213e64345057c2","subject":"Enable non-voting OpenAPI build job","status":"MERGED","created":"2024-06-26 16:51:00.000000000","updated":"2024-07-26 16:28:09.000000000","submitted":"2024-07-26 16:25:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922851-openapi","meta_rev_id":"ae71f8e9643a9450805f9c13e37c7b850e6573f2","_number":922851,"virtual_id_number":922851,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-26 16:25:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2024-07-26 09:52:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2024-07-26 14:33:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2024-07-26 14:33:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-26 17:02:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-26 18:27:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-19 14:37:49.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2024-07-26 14:33:49.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"db10a56de6057042dfa54b1a4974c2f87ef9240d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-06-26 16:51:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47d41dbddae1a2692360f3831372c1fc93bd5f2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 17:02:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/080743101ba94dd6aad09f0cf7a3b695\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3d695109aa264cd285af510ceefe1d82 : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f6e4f7901747497f882fc2351966d593 : SUCCESS in 8m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1cae74ad4ace7f0af938ecb69f2a55afb5863e9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 18:27:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd65bd95ba92450b92665a39c6879b8a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b08ead6d69cd4853acfd3e01890edf4b : SUCCESS in 15m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22438b01575e4737b465d56f6ed00b36 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d355b29851dc4c9d81a1898bf169aa1e : SUCCESS in 11m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/965e396b1b764d7b86c4ea92a38fc7c1 : SUCCESS in 9m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e053142d314b440694ef9f40d2ecab2c : RETRY_LIMIT in 2m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1da35b297155400ea47b7e825dff968a : SUCCESS in 11m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/56198908670041ebbf22937f709def77 : SUCCESS in 58m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c94f067983d4df79a647930004d1697 : SUCCESS in 1h 33m 17s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d6ce195e52a64959881eb244054d5293 : SUCCESS in 14m 16s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2697bb1219234c588f8ab72b15e35351 : SUCCESS in 28m 36s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/4a5f02a54d9f401daeaa6aae06e0e514 : SUCCESS in 33m 14s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59fe52672032430bb9c5991f8f749624 : FAILURE in 22m 39s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/66bcd36a58b24abc9867fc764730229f : SUCCESS in 35m 04s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/563a072b0a264c21b670a89fcabbc8ce : SUCCESS in 9m 54s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/34959c89e3c8451997a8e9060c1dfb7e : FAILURE in 16m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15045eb66b10489888f28f29af23d4bf : SUCCESS in 1h 03m 48s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/71c1f0cd0cb046a38f0b35773918a3b4 : FAILURE in 26m 04s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/4596ad89146348ed8d879a3b6307c053 : SUCCESS in 7m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6d903362261dd191cb82740ab0d716a8f99f379","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-07-19 14:37:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3957a192293d93fbd45ac65c5096f22e28bd9659","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-26 09:52:07.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2f153369b547c746488cb3caf9dc3246df96ef26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 10:15:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f86142e6f5334bb3be1a9e59c7fb4eb0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/59536dbc003f42e3b968595399f0c165 : SUCCESS in 14m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/18acdd93669548c9a68d196ba81e7314 : SUCCESS in 21m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9818b942d3c8d1699603771ff80d0e4b2a05228a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 11:23:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/08fde1c66c43410b9fa0377b14b556f8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c602ee59ba3d4437968475395a808bc5 : SUCCESS in 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3312bb0d1234c569eeefa47a137d891 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ea592b4967f489385cea02a6ab15ac1 : SUCCESS in 11m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/249c495dbdd442af9c4bf0cadf4a0932 : SUCCESS in 10m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/00d648b2161d41e2a1444ac9f5c00bd8 : FAILURE in 14m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e60231da38f49dc87a8c5a484f33b59 : SUCCESS in 10m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/f2f708e34552494383dca8ad01b14a3a : SUCCESS in 1h 00m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/33550d9e8e5647768973f95149b5c67e : FAILURE in 1h 20m 46s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ca4d3c0131614318ab7cf0d1df1dab18 : SUCCESS in 17m 52s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/b39fb01b4e014c288b1ab52bfc76a6c8 : FAILURE in 19m 48s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/986ff5cf0ae84902b53697c9a4c881d0 : FAILURE in 17m 34s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/7f0f8ee28e5e42b582073b08e694e476 : SUCCESS in 29m 28s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d5ee7c14839d4a55a3e46de61a5cafaf : SUCCESS in 21m 48s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c60114cf1e1f4850ad6230e15b99ac9b : FAILURE in 23m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89ffcd37e82c4344b52b3b016276a8df : SUCCESS in 39m 57s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/fbfceee5f16b4b0a9ae320ce267c0b3e : SUCCESS in 30m 41s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/39c3e545dee44d0ab2004921adefc08c : SUCCESS in 7m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2c6a7e2a5da48f88efe558dd46b40c1000822c12","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-26 12:10:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b703946a6e3a1118be52786c4324bee8ad39713c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 12:24:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cc33e65138749b29e639f57b5522454\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ee5d6d02b7b642a98a3066efed553a14 : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3d085ba5dc8a473e96f9e222ae5d0dc4 : SUCCESS in 12m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"abb59f7035e52ffbb86737d6785dab7334395470","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 13:30:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e65b8914b73042d5af778298fe8743cc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f958181081440cfa0cc9ed88f0921f5 : SUCCESS in 14m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53f65ab4d56a4f59a76fa5255a8c3a7b : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b1ba27bec914e15b191a45a3eeffe86 : SUCCESS in 12m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f2107833cf4043bf9721294c84c17f2e : SUCCESS in 9m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7ac87307c3d44b05ab62a4cb0e668e92 : FAILURE in 15m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ea4ac35b46d4b93b62ee69c637e39bc : SUCCESS in 11m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/e246feaac7084890a5dac4bc9a6febf9 : SUCCESS in 52m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5ac0c0749d04a54a73f8fb1f0fe96fe : SUCCESS in 1h 18m 47s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8611dccfaa8e471eb1eeb048786fead9 : SUCCESS in 30m 30s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/c4ddcb217ad24e07aec8be1af08c6841 : FAILURE in 17m 35s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f25a8ba9ca54c19aa12801b3bd55585 : FAILURE in 25m 11s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/161bf281641c4629aae8b1fb65928717 : SUCCESS in 24m 08s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/1b4d0a661a5c45448e1003f66373ba24 : SUCCESS in 10m 36s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/1c8bb6e81af64a969736c0fc9e4060d1 : FAILURE in 10m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f5264e8a8804751b37f0435115340a1 : SUCCESS in 57m 46s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/674fa611a8504a44902b61f8c3413d7d : SUCCESS in 38m 21s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/4988896f536f4dd98538518876692c57 : SUCCESS in 5m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c5a37a4d573b11ab8b5d86fa44988af1e952dc7e","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-07-26 14:33:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5294827eeedf54b526e15f8f0fd3474ed014ca6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 14:34:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fb7be97780eb0cdd0cecb977638779eec0ab3ca7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 16:25:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f9cb9790ac94f359d70030962620a4c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e574e40440e249e2899335e6aefde899 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52680498be7348478bd0bafb28d4b35b : SUCCESS in 12m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/218a9c50187a4953945ae6b847863f92 : SUCCESS in 10m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdde220385034fce8a847f042c62162d : SUCCESS in 11m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/b11ce7eb13954dc68e31cc037a8b8eda : SUCCESS in 1h 00m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9fc39fd7035b4fbf9d0f5421a5d79c6f : SUCCESS in 1h 41m 09s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/0d0aa73e43a04e0c89e6e43b79b16d4c : SUCCESS in 32m 01s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/41aca2f764264fe7b52e7ff3c7056ee7 : SUCCESS in 20m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6da1eecdd7c141d985126d7e53592f73 : SUCCESS in 55m 00s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/2dcf94eea2494f678ccca16d4a95c34c : SUCCESS in 30m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1e9b9ca24a24ba7a9959c8f9eef37d5f19c73a4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 16:25:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ae71f8e9643a9450805f9c13e37c7b850e6573f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 16:28:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee018ab9be2a4baaa43ec03bd0878d2f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5c15c72b28849fd8e53810e1a6a3a40 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aa95af38eaad5a53311d030574c5416369292ffb","revisions":{"6d2df897cf3919596e7900122ebf617f04bfb0aa":{"kind":"REWORK","_number":1,"created":"2024-06-26 16:51:00.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/51/922851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/51/922851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/922851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/922851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/922851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/51/922851/1"}}},"commit":{"parents":[{"commit":"bc57ccbc8bc7741daa83cbe624526a90342960bf","subject":"Remove dependency on pytz","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bc57ccbc8bc7741daa83cbe624526a90342960bf"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-06-26 16:48:16.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-06-26 16:50:32.000000000","tz":120},"subject":"Enable non-voting OpenAPI build job","message":"Enable non-voting OpenAPI build job\n\nWe are going to start building OpenAPI using codegenerator directly in\nthe Keystone and move all the schemas into the codebasis. It is going to\nbe very helpful to immediately see whether generation is possible and\nthe OpenAPI spec is valid. Thus adding it as non-voting for now (and it\nmakes no sense of having non-voting job in gate).\n\nChange-Id: Iaf418d69cbb55fbe9dde3f2ec4213e64345057c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6d2df897cf3919596e7900122ebf617f04bfb0aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6d2df897cf3919596e7900122ebf617f04bfb0aa"}]},"branch":"refs/heads/master"},"aa95af38eaad5a53311d030574c5416369292ffb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-26 09:52:07.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/51/922851/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/51/922851/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/922851/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/922851/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/922851/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/51/922851/2"}}},"commit":{"parents":[{"commit":"181dc7b03fec6b1868dc263d9804f6cdc34ddf1b","subject":"Merge \"Blackify the keystone code base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/181dc7b03fec6b1868dc263d9804f6cdc34ddf1b"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-06-26 16:48:16.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-26 09:52:07.000000000","tz":0},"subject":"Enable non-voting OpenAPI build job","message":"Enable non-voting OpenAPI build job\n\nWe are going to start building OpenAPI using codegenerator directly in\nthe Keystone and move all the schemas into the codebasis. It is going to\nbe very helpful to immediately see whether generation is possible and\nthe OpenAPI spec is valid. Thus adding it as non-voting for now (and it\nmakes no sense of having non-voting job in gate).\n\nChange-Id: Iaf418d69cbb55fbe9dde3f2ec4213e64345057c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/aa95af38eaad5a53311d030574c5416369292ffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/aa95af38eaad5a53311d030574c5416369292ffb"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
