)]}'
{"id":"openstack%2Fkeystone~923181","triplet_id":"openstack%2Fkeystone~master~Icc74d6356494d77f044236c508260eddb8728c6a","project":"openstack/keystone","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-10-07 13:53:23.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-10-07 13:53:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icc74d6356494d77f044236c508260eddb8728c6a","subject":"Enable JsonSchema validation for `project`","status":"MERGED","created":"2024-07-01 18:33:14.000000000","updated":"2024-10-07 13:54:29.000000000","submitted":"2024-10-07 13:53:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"923181-openapi","meta_rev_id":"3092e06f0c20661348b07e395bf81b350eb8c7f1","_number":923181,"virtual_id_number":923181,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-07 13:53:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2024-10-04 14:39:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2024-10-07 11:32:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2024-10-07 11:32:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-01 19:12:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-01 21:20:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-03 14:20:02.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2024-10-04 14:39:32.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-10-07 11:32:01.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"a81e6fb8551e3ce83d16a8400211404c48afeb15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-01 18:33:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1ea5fcf4057c7b77daae72a8dbb02854b0743e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 19:12:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ede92fdfaff4ab89938df297bb656e1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/596c8750bbbf46529d56f21966f08fb2 : FAILURE in 4m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b1743163a0b64a709baeba5b8ac048c0 : SUCCESS in 37m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae1434313c303c67d85170e00c6bb664f5bf8164","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-01 19:25:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c2bdf8750a88ae54ed9b3a8f5f7c394ff729b483","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-01 19:26:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4008b8c0bcf0134a9a69d5190bcf3ef3e007b51e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 19:31:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4787a2dbca4f4167a2cf831291d2871d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2536aa5b09e74062876f2c1a77f35d5d : FAILURE in 3m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e038144bca8442cf886b454cc075e583 : FAILURE in 2m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7ae9c7a0576683938b6ca0f7a55bca6a8bb324a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 21:20:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45c25d513b3248d78e034fcfb01b97aa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5408eb62a7b948cfbad7f96b4bcb7329 : SUCCESS in 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97e6ffd6a5d24a98bfddf44ae2d55a16 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/017d75793a494745ad14ea3cd85b1ee7 : SUCCESS in 11m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8a2ebdc5239f4f4f914e4aa1e578ab55 : SUCCESS in 13m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e7a92883d0346dca8b1ff0356ff3256 : RETRY_LIMIT in 1m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f32e214992bf4ccfbde4b21af6ccb7e2 : SUCCESS in 12m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/08b0df3ad4f24b329f9437461d352057 : SUCCESS in 59m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe0d1207dca7477ab9606a7cff5671ef : SUCCESS in 1h 45m 15s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ca551b7aaecb490898447000bad13c90 : SUCCESS in 27m 37s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/f82ac2a40e40456d8f9f1e74776d2882 : SUCCESS in 40m 23s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa6939bf055b41e6973173e36e79476d : FAILURE in 23m 08s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/953fb9c1351a48ba90602159d4e1e02c : SUCCESS in 28m 22s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c99b0c703feb4955991d3dd16049e1f1 : SUCCESS in 17m 10s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/df24f8c02a2f4b6bbf203f87781dbf96 : FAILURE in 25m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a445fed6ff8426c8e6c52fb5f0825ee : SUCCESS in 58m 46s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/92895fabd6914b8484c7e3c0e1d285f2 : FAILURE in 37m 23s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/3899c56ea17b4d339e3aa5b8c82777f7 : POST_FAILURE in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9b078157d10596089e722f1579ad15f6e8a8d8d9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-07-03 14:20:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"93e6dfabaca50809655bd3e0c3e533c0ec48aa6f","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-03 14:33:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"961c3369b8d566276d1fd4d5552726b50544d575","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-03 14:46:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c074233456b5c1f45499e0496fe51698a520f801","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-26 12:20:14.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"19e9c1b89b2fbaeb3fc30835eff55106a2cf6cb1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 12:35:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed1d923b14bf45e085f13c71d0674fda\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f26a7c09688b4badb01d872f5ba913cb : SUCCESS in 12m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3e0de11ceffa4c8c898c7cd6391b20d9 : SUCCESS in 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"25196398bc2f77529fae87088d3d6ca557b503d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 13:43:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a435b2fb6ec4d339193cb95e8089490\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76213d99380f4246a2d0d2b5edf43ac5 : SUCCESS in 16m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eac7a5950cd44e27af8d2ff6c29d352a : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04d38451d39040f6b81ef8239822989f : SUCCESS in 11m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a19ba198c7d478e826697563d3fd2bd : SUCCESS in 9m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c75b20d9f5a41fabcc456b1ccc7c35c : FAILURE in 15m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1045fe3ebe4443b9ae06cdd149ba22c : SUCCESS in 10m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/c565519dee10484a9b3bdc618125ba8a : SUCCESS in 1h 10m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/238739206a884b4ebda2f18f370ec48f : SUCCESS in 1h 21m 32s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/7a7f58707502433d95f206f78f7de17b : SUCCESS in 30m 20s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/c37910ce65ae4360b777ce24f6044ca5 : FAILURE in 20m 06s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4dfa1d7c03b46dcad53b2ec363327f1 : FAILURE in 27m 17s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/8b8a385f29744cb297201934483d4f1b : SUCCESS in 23m 15s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e341ed78a6034c1da485b2e3dea1bb44 : SUCCESS in 10m 34s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/8a0785bb9acf42e686249af16160a755 : FAILURE in 23m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09ae93739efc48178feef47ddf7ddc14 : SUCCESS in 59m 30s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/2027a7fc96ca4559b2cc8e012ebc3090 : SUCCESS in 39m 08s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/609850d77c094545969c3afe3ee45e6d : POST_FAILURE in 6m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"539d0204c1f378e6414b0a2d3fe3b7d133e10ecb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-08-15 12:19:29.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"2d9c222879f9130da194aa7bea16656b6d6a516b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 12:33:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/619487e803294888a7fbe43c17e9257f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bf099b4e6d484686bf7610d50d42545c : FAILURE in 12m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6a1d6acf15ce42b68c5e1f8428f3a35f : FAILURE in 9m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"023892933baa910f93e8e7d7be7d4d101b58a7d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 13:02:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77903af84b7d44ddb2acc86a8011a24e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5433b5ae18424114839541c27dcdb76f : FAILURE in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c352bd6f3f34a4488079fc8cffa497b : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/463f7d5fed974f8893039051c6a879b8 : FAILURE in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/96ad369c5a5a4089b2245dabdb9dc4f7 : FAILURE in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f8cd46487e24abb840b55475a457ff7 : FAILURE in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eca5e49cd2a4528a566f13810084bae : FAILURE in 5m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/8a78f96d0b38454e9feaaa227c0bc706 : FAILURE in 41m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e74853037f04b879aa05073bf8e5646 : FAILURE in 24m 34s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/09fe61ef7a4c49f4b5009db562c28383 : FAILURE in 25m 12s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/caed36950eca48c1b7c3970c4aafe834 : FAILURE in 20m 47s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16b7a9396b5649f78f1e89b6fe11a01a : FAILURE in 25m 08s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/1f7d1dd2ad294cf59378e3a5872743b4 : FAILURE in 23m 03s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d7632df1e6c84dff9996a5ac1f0ea356 : POST_FAILURE in 7m 47s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/d40141aed1b04218aa3bf225db2abbc1 : FAILURE in 12m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f427e5c8fdff44a2ac790328c114160a : FAILURE in 13m 57s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/db13a5d5ba3b4a6688f1d4fb7ae7b924 : FAILURE in 20m 22s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/b717c89f4cd5443d887e7df087c7273c : POST_FAILURE in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"77b16fd66253c3175d65df5f3b247d8d8eef5dcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-08-15 14:19:29.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"321784efc2f255514bbb56a03a0e0d06178a9f71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 15:22:20.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a62953fbc2b24360a21786bacc31c1ba\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c4d459b880584390af179dcfd313e848 : TIMED_OUT in 1h 01m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f41b65a850964926a4c7d8fe85090f2e : SUCCESS in 37m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"860ac74629582a52e78ce6c4897151387211dd16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 15:36:44.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf161aae28bf4d80b99a5a51bae035f6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f86d16f718fe417a91482966b0ee2575 : SUCCESS in 13m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f8a90273c554f6cae7d812e78afc305 : FAILURE in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aacf5b147c714674ab807f23781d2765 : SUCCESS in 12m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/925f729894b64e01ad3ca42bec1ecfce : SUCCESS in 10m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c17bb448a16949f8b910e42b89d4e384 : FAILURE in 11m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09ff06845b1740baabce1d45ee929962 : SUCCESS in 10m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/e15a1475c30745ab8d65df7a0ff63eeb : SUCCESS in 53m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/00448a34b61044d29a26edf3fdf31bdb : FAILURE in 1h 16m 33s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d4aa3289d67d4ec0bc2fb7a55bc3f4c0 : SUCCESS in 18m 00s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/1f95ab1fbca24e378cef70cbd0638f16 : SUCCESS in 24m 04s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bae441ca018a4cb3b1bd5172d03495fa : FAILURE in 27m 02s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/37dd968b936a41999380248d50866303 : SUCCESS in 30m 19s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d9f36d63b57f47419200ad13085d5b0d : SUCCESS in 19m 59s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/4edb292e79ae4098bce33c040ecce88d : FAILURE in 22m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fc16accad13449f8d873cfe212d7b55 : SUCCESS in 59m 10s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/12cf110460bd44f6bc8b9e1b2f7db477 : SUCCESS in 39m 41s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/58db0e0d9a974c78a00fb323a43cc0fe : FAILURE in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"887395f8fa944db0b9ca9fdd9e57038c3c596a01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-08-15 18:27:55.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"afc5e7be717660401a349cc9acf6aeec7130192e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 19:19:31.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f329eff03e6d45cdbd9072c46c44c0ff\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d6d9c270f10144ecb758e65d12e9454e : SUCCESS in 49m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6ca5954fbdb74d18a494f1d16e9bc6a4 : SUCCESS in 42m 46s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"89fdd143d689e6ee4ab4599948327eaf164512e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 20:11:43.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7a7f81f83514599817a2ee8b21b8293\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34a14e333e634b648d9a501ff08fa3cf : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90f0aac5266544eaa8d746eb3c2b9d75 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f9285cfb6034dac98f97fe487d2685d : SUCCESS in 11m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d848fcc08c84d2b902619a1d570ee41 : SUCCESS in 8m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/116f0be73c014c14bbb13a0cd470901c : FAILURE in 10m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf07dcbf186b42798cb9c8c0ecc9e18a : SUCCESS in 12m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/d4d531a2ca6e42d5a80fbd39615a7338 : SUCCESS in 1h 10m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9375c5226bfc4d799d66b52ab434590a : SUCCESS in 1h 43m 08s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/c61a5cc876a344cb94b5346a5a5c44ac : SUCCESS in 29m 49s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/2bab03fbfda34f91902e5332845f77e5 : SUCCESS in 33m 12s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f67e5f1f70ec486789d37b74e4659db2 : FAILURE in 18m 17s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/7d4be8bad34949a4933c51d2ea9c6dee : SUCCESS in 23m 05s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/7b84bafcc25e4f80acc3e014b4b583e8 : SUCCESS in 12m 52s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ec32e12ac366476383b2527e0deab121 : FAILURE in 15m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24ee16e7e12b413283748ab1bf9c29e4 : SUCCESS in 45m 43s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/fa363bc6616c4f679aeb0227134fb79f : SUCCESS in 31m 16s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/3454fdbf9d664e7baa5070e9c78a285e : FAILURE in 5m 21s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4008d1c2af871c6c2a7cf1c88a4f167a002987cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-10-02 16:20:58.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"b9fdf7eae54d68ba10d55d9cc689cb95a81d7db0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 18:09:19.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b2888b879b4c4de9892bb257ef8e3b3c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62da2e84a6014d2aaafff5a5c266b2f3 : SUCCESS in 14m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/301a7a8f139943ca97a2c4367e3667fe : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d9c0c1597374b759200d185f23bedff : SUCCESS in 12m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af8c7e84cb5f4db5bcc158bf9a39623d : SUCCESS in 9m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/788835722da54ad99c3e49697c432205 : SUCCESS in 11m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/d3fc2c7461bd414784b9bcf33632a2c7 : SUCCESS in 38m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7cd76ac44ded440fa09bace44ea12f50 : SUCCESS in 1h 39m 27s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/647b92aecbfb483696d1509453a84600 : SUCCESS in 35m 42s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/2dfc919e3ddb4141943aac3ca6af4489 : SUCCESS in 43m 06s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ed60e51c2ba4704ae8c8c61f302658c : FAILURE in 29m 38s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/868b7090c4374d6f83efa337549fc16e : SUCCESS in 16m 10s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/76921f9c287a4de59875ecbb37de940c : SUCCESS in 10m 52s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/583b14e175544c4f87c14b7c2c03242b : FAILURE in 13m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0e9ff4adc3241e19e42e67ed1b361a2 : FAILURE in 52m 19s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/6ac6c003d1ad4cc2b345b33a557b1ea0 : SUCCESS in 28m 32s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/6ab68d3968234a4c88142a21a354172c : SUCCESS in 5m 54s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"bfabc0156157477f679fe0df74638a93ef5c8634","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 21:08:35.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73a625e1fc7c4df7b9d5b15b335dc6ab\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4891fa055b8a499eb626be896e182069 : SUCCESS in 36m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5558a7baae12488aae314f3bc1b82744 : SUCCESS in 33m 48s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c2f53402d12fa73a75d2eb89d1c3348adf2db0f2","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-10-04 10:04:45.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"83255cdc32bdd04bdcf46814de2dbbe2b0d3eb25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 11:36:50.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d255865ded3a40f485990b97ffd4e3e4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05c1d7097f8b4f36ba29e8aa2bc1149a : SUCCESS in 14m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0424aca2fb0f483dbb0636a71092ee2b : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/935b1ee331154f8fb2aec938ea5e708b : SUCCESS in 11m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f01a2336ace4e56a8423b349e6dacb8 : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7747479204b54b8cb99cbb3657093f67 : SUCCESS in 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/9ec872af87db4b0d9d35a2caa3a34d41 : SUCCESS in 40m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b35dac3120da4eb7bf06697577697c4c : SUCCESS in 1h 31m 00s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/cc7f058f67794499a574def13df6c39f : SUCCESS in 16m 57s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/7a3dd79acbd24458bc5618734a0bc685 : SUCCESS in 38m 43s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dac413ab0e46499d80998e8e0b0ca66b : FAILURE in 27m 23s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/d87c5f6480d7495aa843925ffcd224eb : SUCCESS in 29m 12s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/1b4517d6ac714ecfb0747ef2bd2eef04 : SUCCESS in 18m 32s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/283fd3938bf84b7b8c456b6a365a52e0 : FAILURE in 21m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0c009e275d64766affff6d0f6992811 : SUCCESS in 51m 46s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/cfa5e67f5e724b19855ced49bea7f0a0 : SUCCESS in 44m 08s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/2a409dada8a8492d9300353c202f8c10 : SUCCESS in 5m 27s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6a6ca2a463384863d09ddfcfbb053fb6a8d1fffa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 12:27:41.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e1436fa148743de96b0be34abbcc6b1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5732ed594419472abdbfe62fdc009ec7 : TIMED_OUT in 1h 00m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c2d81ba0f7964ff78234912524df183d : TIMED_OUT in 1h 00m 54s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f0032657ef5fcbbad66e4207b396ff7a7f79bc39","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-10-04 14:39:32.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"707a3e3e658f63761acb7f3c136855912b10b87b","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-10-07 11:32:01.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"d0bab86682a8bcbe86225c5cb765bfefcfe0a75f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 11:32:37.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"4927c0f29f4b66bd629dcd05cac0b3c4729b96f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 13:53:23.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bb43ec2e2a4408ca80b930b6fba5313\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/797e5215442d47df8933184c71082f0e : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f960a8b081ce43fc8dc1b38c637eba15 : SUCCESS in 12m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0dd5f2019f5a4e609d48a7f618475fde : SUCCESS in 12m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a9db25ea5f4922b7aca00923cfced4 : SUCCESS in 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/a4d0f0c22de34a1787572cc3fab34bff : SUCCESS in 50m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f81506eb68db4b2ca1374356e765fc32 : SUCCESS in 1h 24m 20s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d9d34fae7ef947aab231d3c493ff5400 : SUCCESS in 20m 03s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/efe2371f8f1e45558b8f25e10f55d43b : SUCCESS in 19m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b21872a4ca4407ca734a9fef6f6c9f0 : SUCCESS in 49m 15s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/cffda033dad84993a6bd8bb1433b219f : SUCCESS in 29m 48s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"aa1c56491f8899d42ae790a3d5d4693131b0ebc4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 13:53:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"3092e06f0c20661348b07e395bf81b350eb8c7f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 13:54:29.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef3ab9219bf74241a6274a06bf19b9b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e458ef9bb5ed4ff486156adc310d98c7 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"5cf620a088b5e5d97f5eff01f797a8e2c8eb81dc","revisions":{"ad331f21eaac7cb78b048c4e6fcb22be6820d62a":{"kind":"REWORK","_number":1,"created":"2024-07-01 18:33:14.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/81/923181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/81/923181/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/81/923181/1"}}},"commit":{"parents":[{"commit":"5540296a6e9350f5ccfe4f2069418e403f6b00bc","subject":"Add JsonSchema validation framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5540296a6e9350f5ccfe4f2069418e403f6b00bc"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-01 18:24:36.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-01 18:24:36.000000000","tz":120},"subject":"WIP: Enable JsonSchema validation for `project`","message":"WIP: Enable JsonSchema validation for `project`\n\nAdd JsonSchema validation for `project` resource.\n\n- extend schemas\n- remove lazy validation in the code\n- allow `False`/`True` as query parameters (is_domain test explicitly\n  validates those are valid so it is inconsistent to forbid those. This\n  also matches behavior of other services.\n\nChange-Id: Icc74d6356494d77f044236c508260eddb8728c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ad331f21eaac7cb78b048c4e6fcb22be6820d62a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ad331f21eaac7cb78b048c4e6fcb22be6820d62a"}]},"branch":"refs/heads/master"},"aad940488da42fe09d77e75ae6e2766d2c5045ed":{"kind":"REWORK","_number":2,"created":"2024-07-01 19:25:17.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/81/923181/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/81/923181/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/81/923181/2"}}},"commit":{"parents":[{"commit":"5540296a6e9350f5ccfe4f2069418e403f6b00bc","subject":"Add JsonSchema validation framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5540296a6e9350f5ccfe4f2069418e403f6b00bc"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-01 18:24:36.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-01 19:24:51.000000000","tz":120},"subject":"WIP: Enable JsonSchema validation for `project`","message":"WIP: Enable JsonSchema validation for `project`\n\nAdd JsonSchema validation for `project` resource.\n\n- extend schemas\n- remove lazy validation in the code\n- allow `False`/`True` as query parameters (is_domain test explicitly\n  validates those are valid so it is inconsistent to forbid those. This\n  also matches behavior of other services.\n\nChange-Id: Icc74d6356494d77f044236c508260eddb8728c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/aad940488da42fe09d77e75ae6e2766d2c5045ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/aad940488da42fe09d77e75ae6e2766d2c5045ed"}]},"branch":"refs/heads/master"},"696eb3768f907608a03b60418e78c9a1bdfe831b":{"kind":"REWORK","_number":3,"created":"2024-07-01 19:26:45.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/81/923181/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/81/923181/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/81/923181/3"}}},"commit":{"parents":[{"commit":"5540296a6e9350f5ccfe4f2069418e403f6b00bc","subject":"Add JsonSchema validation framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5540296a6e9350f5ccfe4f2069418e403f6b00bc"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-01 18:24:36.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-01 19:26:35.000000000","tz":120},"subject":"WIP: Enable JsonSchema validation for `project`","message":"WIP: Enable JsonSchema validation for `project`\n\nAdd JsonSchema validation for `project` resource.\n\n- extend schemas\n- remove lazy validation in the code\n- allow `False`/`True` as query parameters (is_domain test explicitly\n  validates those are valid so it is inconsistent to forbid those. This\n  also matches behavior of other services.\n\nChange-Id: Icc74d6356494d77f044236c508260eddb8728c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/696eb3768f907608a03b60418e78c9a1bdfe831b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/696eb3768f907608a03b60418e78c9a1bdfe831b"}]},"branch":"refs/heads/master"},"de3cfe6483e46acf88dc7409da5bbf242f7ce9cb":{"kind":"REWORK","_number":4,"created":"2024-07-26 12:20:14.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/81/923181/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/81/923181/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/81/923181/4"}}},"commit":{"parents":[{"commit":"4bfaaf0f8dbe222458dd7f9ec658a11494dcbbf9","subject":"Add JsonSchema to `credentials`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4bfaaf0f8dbe222458dd7f9ec658a11494dcbbf9"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-01 18:24:36.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-26 12:19:41.000000000","tz":120},"subject":"Enable JsonSchema validation for `project`","message":"Enable JsonSchema validation for `project`\n\nAdd JsonSchema validation for `project` resource.\n\n- extend schemas\n- remove lazy validation in the code\n- allow `False`/`True` as query parameters (is_domain test explicitly\n  validates those are valid so it is inconsistent to forbid those. This\n  also matches behavior of other services.\n\nChange-Id: Icc74d6356494d77f044236c508260eddb8728c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/de3cfe6483e46acf88dc7409da5bbf242f7ce9cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/de3cfe6483e46acf88dc7409da5bbf242f7ce9cb"}]},"branch":"refs/heads/master"},"99c543c6afd638833c24641754602d261dff9640":{"kind":"REWORK","_number":5,"created":"2024-08-15 12:19:29.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/81/923181/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/81/923181/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/81/923181/5"}}},"commit":{"parents":[{"commit":"1e2a32ba866fc3270cc297138506e3228b997990","subject":"Prevent circlular imports in keystone.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1e2a32ba866fc3270cc297138506e3228b997990"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-01 18:24:36.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-08-15 12:19:20.000000000","tz":120},"subject":"Enable JsonSchema validation for `project`","message":"Enable JsonSchema validation for `project`\n\nAdd JsonSchema validation for `project` resource.\n\n- extend schemas\n- remove lazy validation in the code\n- allow `False`/`True` as query parameters (is_domain test explicitly\n  validates those are valid so it is inconsistent to forbid those. This\n  also matches behavior of other services.\n\nChange-Id: Icc74d6356494d77f044236c508260eddb8728c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/99c543c6afd638833c24641754602d261dff9640"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/99c543c6afd638833c24641754602d261dff9640"}]},"branch":"refs/heads/master"},"e4a0c0a05cd4cb61bc514129f0b82a7d8b334311":{"kind":"REWORK","_number":6,"created":"2024-08-15 14:19:29.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/81/923181/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/81/923181/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/81/923181/6"}}},"commit":{"parents":[{"commit":"ea94a9abca4544a36891f60d554674455d454579","subject":"Prevent circlular imports in keystone.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ea94a9abca4544a36891f60d554674455d454579"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-01 18:24:36.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-08-15 14:19:15.000000000","tz":120},"subject":"Enable JsonSchema validation for `project`","message":"Enable JsonSchema validation for `project`\n\nAdd JsonSchema validation for `project` resource.\n\n- extend schemas\n- remove lazy validation in the code\n- allow `False`/`True` as query parameters (is_domain test explicitly\n  validates those are valid so it is inconsistent to forbid those. This\n  also matches behavior of other services.\n\nChange-Id: Icc74d6356494d77f044236c508260eddb8728c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e4a0c0a05cd4cb61bc514129f0b82a7d8b334311"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e4a0c0a05cd4cb61bc514129f0b82a7d8b334311"}]},"branch":"refs/heads/master"},"3ea68fd069f8c7e2f15ae6a457b10c391885c994":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-08-15 18:27:55.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/81/923181/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/81/923181/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/81/923181/7"}}},"commit":{"parents":[{"commit":"19b1f846b7927491ee6df262027f34a4b2224c70","subject":"Prevent circlular imports in keystone.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/19b1f846b7927491ee6df262027f34a4b2224c70"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-01 18:24:36.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-08-15 18:27:55.000000000","tz":0},"subject":"Enable JsonSchema validation for `project`","message":"Enable JsonSchema validation for `project`\n\nAdd JsonSchema validation for `project` resource.\n\n- extend schemas\n- remove lazy validation in the code\n- allow `False`/`True` as query parameters (is_domain test explicitly\n  validates those are valid so it is inconsistent to forbid those. This\n  also matches behavior of other services.\n\nChange-Id: Icc74d6356494d77f044236c508260eddb8728c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3ea68fd069f8c7e2f15ae6a457b10c391885c994"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3ea68fd069f8c7e2f15ae6a457b10c391885c994"}]},"branch":"refs/heads/master","description":"Rebase"},"5cf620a088b5e5d97f5eff01f797a8e2c8eb81dc":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-10-02 16:20:58.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/81/923181/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/81/923181/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/923181/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/81/923181/8"}}},"commit":{"parents":[{"commit":"9acac37b2b2005972ca2a9cbec3aae53bc419300","subject":"Merge \"Prevent circlular imports in keystone.api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9acac37b2b2005972ca2a9cbec3aae53bc419300"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-01 18:24:36.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-10-02 16:20:58.000000000","tz":0},"subject":"Enable JsonSchema validation for `project`","message":"Enable JsonSchema validation for `project`\n\nAdd JsonSchema validation for `project` resource.\n\n- extend schemas\n- remove lazy validation in the code\n- allow `False`/`True` as query parameters (is_domain test explicitly\n  validates those are valid so it is inconsistent to forbid those. This\n  also matches behavior of other services.\n\nChange-Id: Icc74d6356494d77f044236c508260eddb8728c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5cf620a088b5e5d97f5eff01f797a8e2c8eb81dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5cf620a088b5e5d97f5eff01f797a8e2c8eb81dc"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
