)]}'
{"id":"openstack%2Fkeystone~923324","triplet_id":"openstack%2Fkeystone~master~I00ed8eebe411635844bfe7e365ba82984f17d4a9","project":"openstack/keystone","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"35125":{"account":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"last_update":"2024-10-04 16:40:18.000000000","reason":"Change was submitted"},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-10-04 14:43:53.000000000","reason":"\u003cGERRIT_ACCOUNT_7973\u003e replied on the change","reason_account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2024-10-04 14:42:26.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2024-10-04 16:40:18.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-10-04 16:40:18.000000000","reason":"Change was submitted"},"35548":{"account":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"last_update":"2024-10-04 16:40:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I00ed8eebe411635844bfe7e365ba82984f17d4a9","subject":"Add JsonSchema to `credentials`","status":"MERGED","created":"2024-07-02 19:17:05.000000000","updated":"2024-10-04 16:40:18.000000000","submitted":"2024-10-04 16:40:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923324-openapi","meta_rev_id":"89136134c0857bd2dd4f33546f6cf2f396ca6585","_number":923324,"virtual_id_number":923324,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-04 16:40:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":0,"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":2,"date":"2024-10-04 14:43:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2024-10-04 14:42:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":0,"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":1,"date":"2024-10-04 14:43:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"}],"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 19:27:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 21:03:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-19 13:29:15.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"CC"},{"updated":"2024-08-01 15:46:16.000000000","updated_by":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"reviewer":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"state":"CC"},{"updated":"2024-09-27 14:27:40.000000000","updated_by":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"reviewer":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"state":"REVIEWER"},{"updated":"2024-09-27 14:28:47.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-09-27 14:29:20.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"CC"},{"updated":"2024-10-04 14:43:53.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"59b2a7b42a268b45bf2604d38e656aca6b3c6d46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-02 19:17:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fbb728a34098f247252ad2b77676b6762c37ddd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 19:27:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54332917340943f5abc89a4b9d0ee46e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66ace457775d469f965c4d870000a20a : FAILURE in 8m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e307c68651ce4a6ca1c54c28a032ed2b : FAILURE in 8m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4dab0ed8596848ed17092273fb5938262feb688e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 21:03:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bbb64b3944b4972a2c25ebe36aff514\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b2b13a73f664a4aa5e0fe5532e036b6 : FAILURE in 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efe143f28bfa4fc490df7e77eb95394e : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ee296390cee4cbea7c2a37c0103cc17 : FAILURE in 11m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/16cf298a9c5347ceaad2f256f23678d2 : FAILURE in 11m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/98cb35c0ca3a4b2e80b03ac08c7dba87 : RETRY_LIMIT in 1m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/216dbb413693492c9878334788a70429 : SUCCESS in 12m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/1816bae811924220aa83e5bf24844851 : SUCCESS in 1h 05m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/470e26f1818841d19a3e2b3e285ae4ad : SUCCESS in 1h 39m 57s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1ae35e6771b342fc8e2d762d02fb9b90 : SUCCESS in 25m 18s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/a45616b1f8284912a6ec8a803d563206 : SUCCESS in 30m 13s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc66ab714f1b434cb84cd5850c460a78 : FAILURE in 25m 24s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/689754797bd24652a8445733c77d44de : SUCCESS in 30m 47s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/72cf08c2d08a4acc8e548b01e587ab21 : FAILURE in 22m 26s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c038451bc01f4a4a8c5bb34afdd12ee3 : FAILURE in 13m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c82bbc07c5d345358f651dd28866615a : SUCCESS in 51m 56s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/f279faddff0541ee841784afe06e6548 : FAILURE in 36m 07s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/0725244feed543c1a6c37887726c2a38 : POST_FAILURE in 9m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"313374d9a7b587c40e56698c2550453df0e319da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-03 19:46:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c47e19f43064a0ee6becf92a1cbb150fe3397163","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 20:49:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dc791fbfca345cb93c35e89bc2deb60\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b0b386984fc5479cb455916f703fb75e : TIMED_OUT in 1h 01m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/39b3dd9ad3f04da896a3b35474c16b8c : FAILURE in 13m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e90165d105159b22b3af5e1e27fa9ec59056b8d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 21:46:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac93082eeaf241afa875af4896f85e05\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/649f977ecfa042c9acfdcd13dbeba2fa : FAILURE in 13m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e28a97e39968484dac62e624c7315b1a : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/019c665f806f4c0595ff63fe5338b666 : FAILURE in 12m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b92dd86800c743918e7ba17b18c5d8e0 : FAILURE in 12m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d9f548e2319f4a84968a892dcccdb332 : RETRY_LIMIT in 1m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2227c2c610242d7aeab4c2583aa4da9 : SUCCESS in 11m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/20f7e3325ee240f49b469d9c90a4c2ec : SUCCESS in 1h 07m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/271146966e1a4ebf9f51b81ceaac7288 : SUCCESS in 1h 59m 16s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/cc570e7d5b1d404d88c07b311db1565f : SUCCESS in 20m 44s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/06a77b952f5f4926976f84a5ea46e38a : SUCCESS in 40m 25s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/706d90e29c7c4ba6be6dd988f7ad4718 : FAILURE in 24m 05s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/3850aaf9c2dd4d8fb973a41669e8566f : SUCCESS in 25m 19s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/453c3ab07d9d496ebba7710dce1bbd22 : SUCCESS in 18m 57s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c8d072b42212461bbaf0162934c46bf2 : FAILURE in 24m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bba88b4e746b44e9968fba5cc55a85dc : SUCCESS in 52m 56s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/77ba5f33f84149188302d6c6fdb81c63 : SUCCESS in 33m 44s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/c5c115cc105947ddb33411d7b69cc951 : POST_FAILURE in 5m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a2c087e5823dc84cb00f8e50a9f33285832f3f05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-04 07:01:14.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"dd846be4293469f60bdfa567d6159ecac55d79ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-04 07:02:46.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c262ffc5cd1db21dd3b4fc93e7d0836b88f55131","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 07:28:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52969cefa53d4fb8a6e52fde681083c0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b4079638698b4e8a90295a31d3c6e10e : SUCCESS in 15m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d28c8ab6b36747cbb1174079c9eb955e : SUCCESS in 23m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f9aaf9968c8d6258df1a810df9ecd1216b0f8483","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 08:50:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4785b871432c4e57845353bda42cc161\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4b1c2ecd8404abd9ee3143b28b17364 : SUCCESS in 14m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fb6ca6e0f524c1ca68490c441bdeb11 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9b6f587e4e645eda28e126f10128ea8 : SUCCESS in 11m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c63a1bf76db04c128869ecff54ca8170 : SUCCESS in 10m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f6960f00edc2465999da543c7860d6db : RETRY_LIMIT in 1m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27fd72b6d29848edba347a0fc41df6ac : SUCCESS in 15m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/f247b98726d945d7b3e147b2948560bb : SUCCESS in 48m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f6c30f409ed240d4bf3e514652463229 : SUCCESS in 1h 41m 36s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/85630f3e416647059a5fcee6601bf027 : SUCCESS in 30m 23s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/b8c0cbcd3f89457682e3cfddea77a845 : SUCCESS in 40m 37s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/956a0fc520234574b9e7b51adab529aa : FAILURE in 16m 54s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/563ca68893104aa7aa3273913136dcf8 : SUCCESS in 18m 56s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/2dd512898ef447448bf5c9d7b33a0e4d : SUCCESS in 18m 37s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/21768a1b215042148210da13db4a089f : FAILURE in 11m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7170f4fe685945b38668ee7b425b0706 : SUCCESS in 55m 52s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/dac5dd92b9a3422f9f723dcd9fa39e9e : SUCCESS in 40m 43s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/c385e8def1fb4441b090ade7c23abe94 : SUCCESS in 5m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3c3fe064cd66e2fa31a821887462575d6c13ada6","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-07-19 13:29:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7d1d04bd96bc9ed0f286216385ffdd13073a6ced","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-24 15:41:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d1dbf863ee586cf24df43106f4c70fc214c17783","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-26 10:13:12.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"4605335b739489fd84ff922a17731f71fa6a6762","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 10:26:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/217cdf58d2994afb829a4c723e1516c6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0a9fc56ee1f476cab0cd6915a380c1a : SUCCESS in 9m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/02a3c042d8604de789452ed5dc9bac31 : SUCCESS in 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d072a79a9c3cee9c9e0703eb41b538c502a5be95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 11:46:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78e00919ab1b496c9a46c415bd19bf5e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cfa52b80961423688190aa246be3f89 : SUCCESS in 13m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1866d10a63b448da2908db5a22770dc : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d633e8db7d94170975d9340a8d8888a : SUCCESS in 12m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dcce3cc0060a4aa88722534e225fbfc8 : SUCCESS in 10m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55f1902748b542ef91f71bea03a55df3 : FAILURE in 13m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65a96ae0be96454dba2c8c154789981b : SUCCESS in 11m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/68812b0ef59046348c8d6c2a1c71e65f : SUCCESS in 57m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/de6f3ffb3116455db0435f31f3095495 : SUCCESS in 1h 29m 22s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ac44bbe809194c1c8457101e90a49713 : SUCCESS in 31m 34s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/bc96a938555d4e1c829b75efe2fcca5e : FAILURE in 18m 50s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64bcde05f0254026aa6e493763dbdbd3 : FAILURE in 19m 27s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/f397b468ed1c4f17b9712cdda0987ba9 : SUCCESS in 22m 38s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/8ec73003928f4e9a94b04217ce1bbcdd : SUCCESS in 13m 03s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/71932e6788cc48e1a699375ea3f83a60 : FAILURE in 22m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9416bf29d2fc4c37830dd6f2cacf2aab : SUCCESS in 58m 27s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/eef62afa90ef4582bdcb3c57621691f3 : SUCCESS in 38m 38s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/abc7086412354f14a00451be7236003b : SUCCESS in 5m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5c810ad9c71bc69c56339d17683d5de45c5e8d5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-27 11:11:31.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"5ac85856badd4f5b9595cf10aea138983febb2fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-27 11:31:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ba03ab118f34b2b8c8a91b80d24ebb7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce402c2bd4e445f89669582c68cdd47f : SUCCESS in 15m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4820943bd95c450685a2950de66fd8f3 : SUCCESS in 10m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c505045eb701f5e94b974e6d2ee3cfc140b0bb82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-27 12:47:32.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb3389d05e3940399c6ec1db58c2dafc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4e24fce855d4e03bd0440b7b0538da2 : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10eaa5ed34554e52b614082988924243 : FAILURE in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f112f878e7fb4c6cb2d853f887119fd2 : SUCCESS in 9m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f809fa477cd4fe99929d06e4e10bb96 : SUCCESS in 10m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/866fe6e3fdc54699844a7eef8206f921 : FAILURE in 11m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/606c44730d3d4369b5dd41fee96fcd8d : SUCCESS in 12m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/436b46cbd5d64ef0a134f216bec7c6a3 : SUCCESS in 39m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08517d630d06463f81b0505ec86865a7 : SUCCESS in 1h 22m 39s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d093bfcb18a442f6a114a85040642d2b : SUCCESS in 15m 49s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/ffb15c81598148ca99ee5e287baf1275 : FAILURE in 9m 52s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c244dad49ca2464ab2a50015ea8c9cd6 : FAILURE in 24m 53s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/ba14239619c947e996ecdfdf6d416075 : SUCCESS in 23m 46s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/00528053fd0f415bbba43f87ff1b380f : SUCCESS in 19m 24s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/b3ca7a6ca01d404e8cd0b4aedca5f1a1 : FAILURE in 23m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69d89be5a8df4507863ab34c652fd556 : SUCCESS in 37m 24s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/8d10ef798ed94bdfbba64d5257b57dbb : SUCCESS in 38m 19s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/66726644b3cc4e69a3dfdfeae32218ee : SUCCESS in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b8bd668b8955848709eba58b168779c896cc484b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-29 15:45:14.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"fa6a77538d072284a12519855091ab7e927b9fdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 17:23:40.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71190d51d1a14f99a367d1c9c4258e71\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee090811d46a433cb80a287a1ca677dd : SUCCESS in 11m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d6167aaf72c473681436a8d60e5e8c3 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3de3f974d80b4ac4bba84ad87f92ade4 : SUCCESS in 11m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52fbf04d8ca64777872f8942781aaf42 : SUCCESS in 10m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c4741ac46015434aa57df0d2887c27f2 : FAILURE in 15m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67916c62be354681bc9e18679e2ac920 : SUCCESS in 12m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/dae528a1836b481bb3cb21d599472482 : SUCCESS in 46m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/823b729d1418409eb98b883d6c2f9652 : SUCCESS in 1h 32m 59s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/01545b1873f54662af08ccb7d623fe79 : SUCCESS in 32m 44s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/922d896872064259891085bae8bd38e0 : FAILURE in 10m 37s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12b8e4ce8d114048a5b36466a978f7f1 : FAILURE in 15m 29s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/2583c4407bd44b80b5a187d84c18ce2c : SUCCESS in 19m 51s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/89d9d863edfa47c7b1a5172775bf5fe7 : SUCCESS in 11m 01s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/cc463fd07aa74882a099a53638870c7b : FAILURE in 14m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2ea99cf9ff84b4a9c7023f0528abfb1 : SUCCESS in 1h 05m 37s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/3ef8f8bd88444867b3a1a8187d81faae : SUCCESS in 37m 54s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/abe919e6e4254f6a9dd10687b95c4bcf : SUCCESS in 5m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"673318402ce4cb06dc7b33cffdcf3ef674f745d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 17:39:26.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ff44274e3ef4d948de8bd74f6a1c0e1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9cf601032288461a8ae9a41ccc3adfbc : SUCCESS in 51m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/60f9ac5cef564bb591394dd07cddbee1 : SUCCESS in 54m 13s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"658da727f6d831b1e10c04c06cf6d7a8e3082274","author":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"date":"2024-08-01 15:46:16.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b4a45019fa2157526b222aa3e509d68e1dac1463","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-08-01 15:48:30.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"40ef0784bfa7ce6de7cb0fb9fa641237df093e0a","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-08-01 15:48:46.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"efc42550f090e9d1e667d95672dd4205d49865a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 17:02:21.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97e299ac57534409be5a64e4322b50d5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0eba5f3a030640c4ae39d1d45b4d7cd8 : SUCCESS in 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb6985551fc048ab9a23b532b04f4907 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38f80eb7e58c4b8d87d9290d522e077e : SUCCESS in 12m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c83430c401174afcadb703e9c4c93218 : SUCCESS in 9m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a7ccc56c2bbe4cfca1d61087121a03cd : FAILURE in 15m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dd5db56238643f592351d4b23016601 : SUCCESS in 11m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/a1f54540c93449f6ae7c33b18d2bfd62 : SUCCESS in 58m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ef9cf4bfa164966a464e20b3c358ecb : SUCCESS in 1h 11m 05s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/3f382a4934e64dbbbd168bbf86b7c614 : SUCCESS in 18m 34s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/2af50335568d404b93f709d18c55d1f5 : FAILURE in 18m 39s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a451a4ab9fa4e86b2eacc397acb50d6 : FAILURE in 25m 25s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/3fca387f50204c55a4b33d15e935c851 : SUCCESS in 27m 47s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/975d551f7d6c4fad8888148183dde2d5 : SUCCESS in 20m 23s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/1bf25e31fc1e484591410917ad0aecd6 : FAILURE in 20m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8bbb1399e52f48e4bb276afab302e697 : SUCCESS in 54m 55s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/254b2d55c74443f98b033407709e2469 : SUCCESS in 35m 50s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/68502761b5614c8b8fd202fde1d2a660 : SUCCESS in 5m 54s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"dd1ac8852a0eb8d099d0f9c806e52ff20f121c0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 17:24:39.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/477014c581684ebbbf00246f5cb8a225\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3dcea9001ff04cb19c09e82d89c55241 : SUCCESS in 1h 00m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8770a823d2164dd7903854fd9067b0f5 : SUCCESS in 52m 08s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9c6bf5eaf6b1cf57d5a613cf527ae8966e221482","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2024-09-27 14:27:40.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"032c281edf3a1461bb2d9eae0d772109f4157649","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-09-27 14:28:47.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"8997a00815bb99aa6a771fc1fb1f548169191716","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-09-27 14:29:20.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7d3d28097419360530d75afaf8c96bcc8f568d25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-09-27 16:01:59.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"027210f687c6a14e445b0a324dfe05628a9f9e99","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-09-27 16:02:25.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"8f0d6bc52b4ca6e2f674dd990cfd0f2e7f98a94e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-09-30 07:05:53.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"e80f5b8b591f751f4b23877b46163ddbae60bcc3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 08:08:34.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8a1001b109f481a96f6c71fc42681c1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/22b1639449ef4df8a9a4b1ba9f5ce15a : SUCCESS in 54m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1348471d90dd4c0da541f46db8fa014b : TIMED_OUT in 1h 01m 01s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"52212eff22944e6b39b8f532d4d43793835c8f3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 08:38:03.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/546d4f6fec7c44a5863d13054a47db2b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b61ea61770ca40319bceeb1546dbfcb1 : SUCCESS in 13m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43169726d313436f947c8c57be8c0f23 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54a8c1e6c6b2471d95a2ca46761f14d3 : SUCCESS in 12m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f452dc404d1042c08971e0208468f16a : SUCCESS in 14m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebaf053b07524813b364d43a01ef026c : SUCCESS in 11m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/36a3020cdf374c8290d14bc23df4a265 : SUCCESS in 55m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e8ddbe119424222ad27cee8a9c17451 : SUCCESS in 1h 31m 05s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/3dbc61e094a745dea3d94c9168422eba : SUCCESS in 23m 15s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/f64e9d41dc424e039381d6ef0b44d951 : SUCCESS in 30m 02s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47435ac8937742879db3d08268c281ef : FAILURE in 13m 16s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/6175815c834a4e1bbce2b4eb59b23000 : SUCCESS in 32m 20s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/0e21da9378ac4384b8e75d7bf31d5dd0 : SUCCESS in 10m 36s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/2ab70989472343ad842688f083b0c3f1 : FAILURE in 13m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c917346e50634607b7531b4c99a4f937 : SUCCESS in 56m 18s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/b27be0d8bb78400dad159b894b39b9ce : SUCCESS in 20m 42s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/86977c6b78b3463a82898c7f6f17fad9 : SUCCESS in 6m 23s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"8f8563ada54415c5506adc53d9c77f0af8eaf950","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-10-04 14:42:26.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"e67dcf2acd7c07d45aa3e54e45bb6867dab43ffc","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-10-04 14:43:53.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"aaa4611d5b14b521f93f469a791bbc847549c03d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 14:44:35.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"e6968040be90c5ff80f8f23eefe41a40c11e7b3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 16:40:17.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bde66b83f36a4c1186affc1ca673b5e1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b27cecda30b440029b9795a4e1fc7c90 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9eba65db2e6d471e8c6bc768b5b667e1 : SUCCESS in 12m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bcf5bc7ed64548769869cb4fa3e547f8 : SUCCESS in 11m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67c561dfba7c4a0f869aef059eb09115 : SUCCESS in 12m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/3319938c97c2472e86c30a3c6518b627 : SUCCESS in 46m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64dbab1d42f341d58d57d8434914ddd0 : SUCCESS in 1h 16m 38s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e3d019eb3d034f8c9e55f4b55c1b1334 : SUCCESS in 21m 00s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/24191bd0812341f8be4a16b21042e558 : SUCCESS in 34m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dad490744f1743949b6cf6bbb28587fd : SUCCESS in 1h 02m 43s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/964d2b280fe54225bc3c082b66907e43 : SUCCESS in 42m 09s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"89136134c0857bd2dd4f33546f6cf2f396ca6585","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 16:40:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"f74ddfdf9d41df382eddbfc9a4076a1289ef1c61","revisions":{"c702fbdf2588d92671da45e4770a30f0013b2676":{"kind":"REWORK","_number":1,"created":"2024-07-02 19:17:05.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/24/923324/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/923324/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/923324/1"}}},"commit":{"parents":[{"commit":"696eb3768f907608a03b60418e78c9a1bdfe831b","subject":"WIP: Enable JsonSchema validation for `project`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/696eb3768f907608a03b60418e78c9a1bdfe831b"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-02 19:16:18.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-02 19:16:18.000000000","tz":120},"subject":"WIP: Add JsonSchema to `credentials`","message":"WIP: Add JsonSchema to `credentials`\n\nAdd json schema validation decorators to the the credential resource.\n\nChange-Id: I00ed8eebe411635844bfe7e365ba82984f17d4a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c702fbdf2588d92671da45e4770a30f0013b2676"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c702fbdf2588d92671da45e4770a30f0013b2676"}]},"branch":"refs/heads/master"},"18e1e9abb0e2d7f9d573e3b52528c2d23aebf317":{"kind":"REWORK","_number":2,"created":"2024-07-03 19:46:07.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/24/923324/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/923324/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/923324/2"}}},"commit":{"parents":[{"commit":"696eb3768f907608a03b60418e78c9a1bdfe831b","subject":"WIP: Enable JsonSchema validation for `project`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/696eb3768f907608a03b60418e78c9a1bdfe831b"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-02 19:16:18.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-03 19:45:48.000000000","tz":120},"subject":"Add JsonSchema to `credentials`","message":"Add JsonSchema to `credentials`\n\nAdd json schema validation decorators to the the credential resource.\n\nChange-Id: I00ed8eebe411635844bfe7e365ba82984f17d4a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/18e1e9abb0e2d7f9d573e3b52528c2d23aebf317"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/18e1e9abb0e2d7f9d573e3b52528c2d23aebf317"}]},"branch":"refs/heads/master"},"e084b1e45160bd344be75febc0c26e2449dde675":{"kind":"REWORK","_number":3,"created":"2024-07-04 07:01:14.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/24/923324/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/923324/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/923324/3"}}},"commit":{"parents":[{"commit":"696eb3768f907608a03b60418e78c9a1bdfe831b","subject":"WIP: Enable JsonSchema validation for `project`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/696eb3768f907608a03b60418e78c9a1bdfe831b"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-02 19:16:18.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-04 07:00:42.000000000","tz":120},"subject":"Add JsonSchema to `credentials`","message":"Add JsonSchema to `credentials`\n\nAdd json schema validation decorators to the the credential resource.\n\nChange-Id: I00ed8eebe411635844bfe7e365ba82984f17d4a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e084b1e45160bd344be75febc0c26e2449dde675"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e084b1e45160bd344be75febc0c26e2449dde675"}]},"branch":"refs/heads/master"},"e3973b5e3e0c485af855d0658944b972679a840f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-07-04 07:02:46.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/24/923324/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/923324/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/923324/4"}}},"commit":{"parents":[{"commit":"5540296a6e9350f5ccfe4f2069418e403f6b00bc","subject":"Add JsonSchema validation framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5540296a6e9350f5ccfe4f2069418e403f6b00bc"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-02 19:16:18.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-04 07:02:46.000000000","tz":0},"subject":"Add JsonSchema to `credentials`","message":"Add JsonSchema to `credentials`\n\nAdd json schema validation decorators to the the credential resource.\n\nChange-Id: I00ed8eebe411635844bfe7e365ba82984f17d4a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e3973b5e3e0c485af855d0658944b972679a840f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e3973b5e3e0c485af855d0658944b972679a840f"}]},"branch":"refs/heads/master","description":"Rebase"},"4bfaaf0f8dbe222458dd7f9ec658a11494dcbbf9":{"kind":"REWORK","_number":5,"created":"2024-07-26 10:13:12.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/24/923324/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/923324/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/923324/5"}}},"commit":{"parents":[{"commit":"0b1e1b2240f80e2a59e624619e45790fe31681e3","subject":"Add JsonSchema validation framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0b1e1b2240f80e2a59e624619e45790fe31681e3"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-02 19:16:18.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-26 10:01:41.000000000","tz":120},"subject":"Add JsonSchema to `credentials`","message":"Add JsonSchema to `credentials`\n\nAdd json schema validation decorators to the the credential resource.\n\nChange-Id: I00ed8eebe411635844bfe7e365ba82984f17d4a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4bfaaf0f8dbe222458dd7f9ec658a11494dcbbf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4bfaaf0f8dbe222458dd7f9ec658a11494dcbbf9"}]},"branch":"refs/heads/master"},"ace586bc296a6b1471ef5283deb5e1964b6ed174":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-07-27 11:11:31.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/24/923324/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/923324/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/923324/6"}}},"commit":{"parents":[{"commit":"e9ababa3ec092e51b3a69186083c991778381c03","subject":"Add JsonSchema validation framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e9ababa3ec092e51b3a69186083c991778381c03"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-02 19:16:18.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-27 11:11:31.000000000","tz":0},"subject":"Add JsonSchema to `credentials`","message":"Add JsonSchema to `credentials`\n\nAdd json schema validation decorators to the the credential resource.\n\nChange-Id: I00ed8eebe411635844bfe7e365ba82984f17d4a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ace586bc296a6b1471ef5283deb5e1964b6ed174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ace586bc296a6b1471ef5283deb5e1964b6ed174"}]},"branch":"refs/heads/master","description":"Rebase"},"be8b79327def5d5844dcae7b1d3517dd7202e5f4":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-07-29 15:45:14.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/24/923324/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/923324/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/923324/7"}}},"commit":{"parents":[{"commit":"9c936969023c8b23a12fed3c5379dd859835bab5","subject":"Add JsonSchema validation framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9c936969023c8b23a12fed3c5379dd859835bab5"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-02 19:16:18.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-29 15:44:36.000000000","tz":120},"subject":"Add JsonSchema to `credentials`","message":"Add JsonSchema to `credentials`\n\nAdd json schema validation decorators to the the credential resource.\n\nChange-Id: I00ed8eebe411635844bfe7e365ba82984f17d4a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/be8b79327def5d5844dcae7b1d3517dd7202e5f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/be8b79327def5d5844dcae7b1d3517dd7202e5f4"}]},"branch":"refs/heads/master"},"5e7ef45106c96cc046138b88f6c61d2754f5977d":{"kind":"REWORK","_number":8,"created":"2024-08-01 15:48:30.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/24/923324/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/923324/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/923324/8"}}},"commit":{"parents":[{"commit":"9c936969023c8b23a12fed3c5379dd859835bab5","subject":"Add JsonSchema validation framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9c936969023c8b23a12fed3c5379dd859835bab5"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-02 19:16:18.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-08-01 15:48:19.000000000","tz":120},"subject":"Add JsonSchema to `credentials`","message":"Add JsonSchema to `credentials`\n\nAdd json schema validation decorators to the the credential resource.\n\nChange-Id: I00ed8eebe411635844bfe7e365ba82984f17d4a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5e7ef45106c96cc046138b88f6c61d2754f5977d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5e7ef45106c96cc046138b88f6c61d2754f5977d"}]},"branch":"refs/heads/master"},"076409f18eb3ac3d85efe284ecc3c5c2624dfcf9":{"kind":"REWORK","_number":9,"created":"2024-09-27 16:01:59.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/24/923324/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/923324/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/923324/9"}}},"commit":{"parents":[{"commit":"9c936969023c8b23a12fed3c5379dd859835bab5","subject":"Add JsonSchema validation framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9c936969023c8b23a12fed3c5379dd859835bab5"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-02 19:16:18.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-09-27 16:01:37.000000000","tz":120},"subject":"Add JsonSchema to `credentials`","message":"Add JsonSchema to `credentials`\n\nAdd json schema validation decorators to the the credential resource.\n\nChange-Id: I00ed8eebe411635844bfe7e365ba82984f17d4a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/076409f18eb3ac3d85efe284ecc3c5c2624dfcf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/076409f18eb3ac3d85efe284ecc3c5c2624dfcf9"}]},"branch":"refs/heads/master"},"f74ddfdf9d41df382eddbfc9a4076a1289ef1c61":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-09-30 07:05:53.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/24/923324/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/923324/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/923324/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/923324/10"}}},"commit":{"parents":[{"commit":"9acac37b2b2005972ca2a9cbec3aae53bc419300","subject":"Merge \"Prevent circlular imports in keystone.api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9acac37b2b2005972ca2a9cbec3aae53bc419300"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-02 19:16:18.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-09-30 07:05:53.000000000","tz":0},"subject":"Add JsonSchema to `credentials`","message":"Add JsonSchema to `credentials`\n\nAdd json schema validation decorators to the the credential resource.\n\nChange-Id: I00ed8eebe411635844bfe7e365ba82984f17d4a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f74ddfdf9d41df382eddbfc9a4076a1289ef1c61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f74ddfdf9d41df382eddbfc9a4076a1289ef1c61"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
