)]}'
{"id":"openstack%2Fkeystone~923991","triplet_id":"openstack%2Fkeystone~master~Ia00f4209cde8f64828dc2d827d49bfc4bd6c1efa","project":"openstack/keystone","branch":"master","topic":"renovate","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-07-19 16:49:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia00f4209cde8f64828dc2d827d49bfc4bd6c1efa","subject":"Add pre-commit","status":"MERGED","created":"2024-07-11 19:44:00.000000000","updated":"2024-07-19 16:49:59.000000000","submitted":"2024-07-19 16:49:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923991-renovate","meta_rev_id":"61e0807f9c6279502c872f704980b9af5caaa0d5","_number":923991,"virtual_id_number":923991,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-19 16:49:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2024-07-12 18:36:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2024-07-19 14:26:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2024-07-19 14:29:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-11 19:50:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-11 21:13:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-12 14:05:08.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-07-19 14:26:31.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"a6bd97358ebc7dd44d6c4897cf5a35b7022518bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-11 19:44:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be0ef3c99287bfd94b8ea1ec67dc1a405b322226","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 19:50:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86107f512a854af9b3ff468ac4f85ac7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b526155df5a48da8d224ac47381cb7b : FAILURE in 4m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/45433832d33d41d78c92cd5f860d335b : FAILURE in 3m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"88293de854b875c23740d3735af254a0e461e384","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 21:13:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9707909b9ed44edd8626dfb11a321e8d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1452a85fef1d40088eadd863babdc9c3 : SUCCESS in 15m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfefee05386243cda4c51eb38084ff21 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f4b59d32e6d496282d548f5222580de : SUCCESS in 15m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c8f5f152f384b33a5b6af0e2f403a8b : SUCCESS in 10m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ef2b35dd1ec44425b090357736a24786 : RETRY_LIMIT in 1m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cef2b5cd6dc458496130947bd74a722 : SUCCESS in 13m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/105faaa94fce4d3f9501e2f8d0747bf6 : SUCCESS in 54m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f7c72eb74fc4482480f9432a87d2ffb3 : SUCCESS in 1h 28m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/340937f51fb34bbc9be35933030fbd77 : SUCCESS in 10m 31s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/4d4bad5e97714d47960c1dd1d7da9fd9 : SUCCESS in 34m 33s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/f772c0ebe5694bc38966c014ac7f7872 : SUCCESS in 32m 12s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c6cf21ce93344cab565e0089848596f : FAILURE in 24m 02s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/e26ea396b2884033960c1a309dd637f8 : SUCCESS in 31m 16s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/b8a3ae4d3a364baa8ab9ae7f38a09640 : SUCCESS in 13m 30s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/5c772fc39db748ec8d6e6b8625d86911 : FAILURE in 22m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2592f9e7e6984f739a880eaf69c9101b : SUCCESS in 1h 04m 23s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/30a7f24fc26e4fa784ad300409f328df : SUCCESS in 40m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a545794b01fde49369d20abbd18a1241acb4825d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 06:48:48.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"fc3ba0aa12d1d23533ed1d1bb33fa44408e975ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 06:59:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fdc9f66fa14474ab5b591b5ffc5763d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/85b32745230c4f4083bc341ee19850bb : FAILURE in 3m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9f507c189de146798636a296bd9e4ed6 : FAILURE in 2m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2f20dc32f1443ef3d9950f136d08b0696e90632e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 08:16:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96d2fbd21a084e54815843ab76763e3e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55ee89871ac7432faab43c1fa3f344b3 : SUCCESS in 15m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f1574b0f1c34d5da2933cdc3f3e88e3 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/166b077bd57147a8aa5ff7781296447f : SUCCESS in 13m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32215cb2f4fa4bdaa953d456a16a022e : SUCCESS in 11m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aaa51b41a870495a81e7781b91de3782 : RETRY_LIMIT in 1m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f06ae2a2acff4fb8bf547dfa3e0ada0a : SUCCESS in 12m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/1a87726930984a919e823e3bf6844ffb : SUCCESS in 1h 00m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/457dd628a16449bd993e7c079f025557 : SUCCESS in 1h 26m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b9bd02b8bde4332b5ad2da361b6c59a : SUCCESS in 10m 26s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8b0fe482d4cf4aec802d6c5f45c485e9 : SUCCESS in 30m 51s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/676ec7ce02b2409e9c32cf2410732773 : SUCCESS in 38m 45s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04cd3403291a44cbb90a2647deb3be68 : FAILURE in 26m 29s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/24ca61fe3ff840d9a1e08347ba26ddd7 : SUCCESS in 21m 11s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/1c5ac589216249f5841e56ca1b3fba82 : SUCCESS in 10m 31s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/6eb1ae3aa8b04fccb22efc62db8576f7 : FAILURE in 20m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48607cba33944829a8aa324c68896c21 : SUCCESS in 51m 46s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/1ba3c5485e6743c6b11ab1036d1056fb : FAILURE in 31m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ca05da3d18908bb38edf6ce8bca354f0843dc1e1","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-07-12 14:05:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1deccc020095acfb8c66d3a09633ee0d7ac3b439","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 18:36:13.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"93f884b3ff9cb9de7229061e7f5e5d4371d2df85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 18:45:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/070afb8fd84e4671b870c4e26d28c98b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dff5c49b8efc4dc5b3c8ec409d92a8dd : FAILURE in 5m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6e53edb4a701420c8065d27b38886339 : FAILURE in 3m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e05b41e666bf8007cf6ef9d4079aceaea0acc56b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 19:58:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60706b92272a4e009d1b5ab788a57050\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39ef51e73ac34b479387217265e91f6c : SUCCESS in 14m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef17b610dfa8437796dbc9ffaf1a40f1 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8320964b2f6e474795303f8495ec8a48 : SUCCESS in 11m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/46dbdbdbbe07451181eaaf1be9b168c6 : SUCCESS in 14m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8a36946c84a848d1a1b543271787bef1 : RETRY_LIMIT in 1m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d62b9ed7fc39475b920562446a257eab : SUCCESS in 13m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/f4eda7ba913d4a799ce2e7b3925ac037 : SUCCESS in 58m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10178d628d4a45968a27f17735481df2 : SUCCESS in 1h 18m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7543950aca0b4e538624cd84c77cab2c : SUCCESS in 10m 42s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2b1516d08ac243498f53ec59bd377743 : SUCCESS in 23m 00s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/0ddd8ad76b0546b39a5429cfebf3e930 : SUCCESS in 35m 53s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/563c8e7e56134fb68e91f684ce4ea893 : FAILURE in 19m 28s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/62c91251305549d3b935090724fd7554 : SUCCESS in 22m 57s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/6e0ab9236fc343f99a113a8d6472170c : SUCCESS in 13m 24s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/783eb179565d41b180aebd0bc2df6ad7 : FAILURE in 14m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0524276d67f14e8a8e7be37d78d075c0 : SUCCESS in 48m 45s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/60daefd5d06a4ed4acccee757295e775 : SUCCESS in 35m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e180adb36ba4180e2c0dfc7d0bad92c22a91df4a","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-07-19 14:26:31.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"71833144eaccc3cfa2cab9cf0796b15446331653","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-07-19 14:29:21.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ed0409e341751876bfe644a1e7e84393f6aa9456","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 14:30:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"88b491d621f65e7491e5dd632ba73b155e001c90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 16:49:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c46fa44544b0493ca070aaa6a54bec35\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3cea96308624d05ab3e3208e960a10b : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ac299648ace49faa2b4d84baa4c6a7a : SUCCESS in 12m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/93b9ea60a6a74c7c9474c6f6cc3067b6 : SUCCESS in 10m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c130395b46074f93a182ad35736fc741 : SUCCESS in 12m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/c7786e65efd14a0e93be71d8981d7d1f : SUCCESS in 1h 03m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/20b3a7dfab9c460e8676efe2ac515b70 : SUCCESS in 1h 35m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ad91938a63e4e1ba7042f08c502a4ca : SUCCESS in 13m 27s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2cdbbeb44e454d1faeef394c182b2663 : SUCCESS in 22m 12s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/6b516752a6be4729ad82d6950fd9988b : SUCCESS in 28m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ed470fdf6ab400cbdbaaeae299867f8 : SUCCESS in 46m 00s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/4c4da3fe9a4c4d1b8f22defb5d1ae08b : SUCCESS in 40m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"61e0807f9c6279502c872f704980b9af5caaa0d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 16:49:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"430bebe376ee76e8b7d0a8cae1e7d65b35816234","revisions":{"2c99a6b42a9ec1eb5f408b8f2ad389dd446db3d1":{"kind":"REWORK","_number":1,"created":"2024-07-11 19:44:00.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/91/923991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/91/923991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/923991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/923991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/923991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/91/923991/1"}}},"commit":{"parents":[{"commit":"0bc2af48d5eabb59627f0b733a9fec42ff3759aa","subject":"Merge \"reno: Update master for unmaintained/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0bc2af48d5eabb59627f0b733a9fec42ff3759aa"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 15:03:28.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 15:03:28.000000000","tz":120},"subject":"Add pre-commit","message":"Add pre-commit\n\nAs a starting point for the code renovation it makes sense to add\npre-commit with the same configuration as we spread across the other\nservices. For now comment out hooks that fail and address them one by\none.\nAt the end of the series we would have pre-commit executing all the\nnecessary checks (linting, flake, black, mypy, etc) and it will be\ninvoked directly in the `tox -e linters` to unify all the processing. We\nare going to re-format source code with `black` adding revisions to\ngit blame ignore to reduce amount of noise. This will help us to have\nreasonable formatting of the code with possibility to just auto-format\nthe code.\n\nChange-Id: Ia00f4209cde8f64828dc2d827d49bfc4bd6c1efa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2c99a6b42a9ec1eb5f408b8f2ad389dd446db3d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2c99a6b42a9ec1eb5f408b8f2ad389dd446db3d1"}]},"branch":"refs/heads/master"},"cbf22f7098660973f5b67c03a4799c56049f8bb5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-07-12 06:48:48.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/91/923991/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/91/923991/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/923991/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/923991/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/923991/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/91/923991/2"}}},"commit":{"parents":[{"commit":"0bc2af48d5eabb59627f0b733a9fec42ff3759aa","subject":"Merge \"reno: Update master for unmaintained/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0bc2af48d5eabb59627f0b733a9fec42ff3759aa"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 15:03:28.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 05:59:40.000000000","tz":120},"subject":"Add pre-commit","message":"Add pre-commit\n\nAs a starting point for the code renovation it makes sense to add\npre-commit with the same configuration as we spread across the other\nservices. For now comment out hooks that fail and address them one by\none.\nAt the end of the series we would have pre-commit executing all the\nnecessary checks (linting, flake, black, mypy, etc) and it will be\ninvoked directly in the `tox -e pep8` to unify all the processing. We\nare going to re-format source code with `black` adding revisions to\ngit blame ignore to reduce amount of noise. This will help us to have\nreasonable formatting of the code with possibility to just auto-format\nthe code.\n\nChange-Id: Ia00f4209cde8f64828dc2d827d49bfc4bd6c1efa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/cbf22f7098660973f5b67c03a4799c56049f8bb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/cbf22f7098660973f5b67c03a4799c56049f8bb5"}]},"branch":"refs/heads/master"},"430bebe376ee76e8b7d0a8cae1e7d65b35816234":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-07-12 18:36:13.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/91/923991/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/91/923991/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/923991/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/923991/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/923991/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/91/923991/3"}}},"commit":{"parents":[{"commit":"1b78b57ec5a80ae6ad72c2ea9bb93734e1c88d31","subject":"Merge \"Improve configuration of out-of-tree identity drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1b78b57ec5a80ae6ad72c2ea9bb93734e1c88d31"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 15:03:28.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 18:30:33.000000000","tz":120},"subject":"Add pre-commit","message":"Add pre-commit\n\nAs a starting point for the code renovation it makes sense to add\npre-commit with the same configuration as we spread across the other\nservices. For now comment out hooks that fail and address them one by\none.\nAt the end of the series we would have pre-commit executing all the\nnecessary checks (linting, flake, black, mypy, etc) and it will be\ninvoked directly in the `tox -e pep8` to unify all the processing. We\nare going to re-format source code with `black` adding revisions to\ngit blame ignore to reduce amount of noise. This will help us to have\nreasonable formatting of the code with possibility to just auto-format\nthe code.\n\nChange-Id: Ia00f4209cde8f64828dc2d827d49bfc4bd6c1efa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/430bebe376ee76e8b7d0a8cae1e7d65b35816234"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/430bebe376ee76e8b7d0a8cae1e7d65b35816234"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
