)]}'
{"id":"openstack%2Fkeystone~923996","triplet_id":"openstack%2Fkeystone~master~I7eca91e5775919cfd78b9825cdd7a9808037b1aa","project":"openstack/keystone","branch":"master","topic":"renovate","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-07-26 07:56:25.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7eca91e5775919cfd78b9825cdd7a9808037b1aa","subject":"Blackify keystone.application_catalog","status":"ABANDONED","created":"2024-07-11 19:44:00.000000000","updated":"2024-07-26 07:56:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6a9f2f48a94e623dd6aee4bf77e50916b1d428dc","_number":923996,"virtual_id_number":923996,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-12 20:18:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-11 20:47:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-11 21:27:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e9c4635c69ceb98e3286d4a0f49bd45a6e9c33a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-11 19:44:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a20632b7830d08d754943a36b4b14d9dd7bea61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 20:47:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5a783e363fb466eb56b9dce46981c2d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c78212b733e94221a67d08e1f7b3e9e1 : TIMED_OUT in 1h 01m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/381549a2019142279e0188cc16e69301 : FAILURE in 3m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"deb7b9ba9b5ff8aee3d6f22798c375d08a02251a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 21:27:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a41b80b6fd8a4625ab29adf469ef66fb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fc1f7adb65543a380ca679840f57415 : SUCCESS in 15m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7b084052474447f9533887f20e4712f : FAILURE in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7427a4fe007b49289fefe11e765d3c73 : SUCCESS in 12m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/91028d8c78b14fe2ac379e2323f91d3f : SUCCESS in 10m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fab0b2706df4475cac236968aaa7b6ff : FAILURE in 15m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b1553c19e334bb8a30762790646eca2 : SUCCESS in 11m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/3ee80b2e7bf04083bfe7183d609f8c27 : SUCCESS in 58m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/959beb1d0a604c42a4abfde39da67b37 : SUCCESS in 1h 35m 30s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/499ac58823ec4c10a90516cf2dde3665 : SUCCESS in 31m 20s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/5c8c324a0d414aeca7c673ee1028a13d : SUCCESS in 46m 28s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6853a712d1c4360bba7f01806c360a3 : FAILURE in 25m 35s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/329999897b5a4db5a33488212ed5c7ba : SUCCESS in 31m 04s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/ddebba284e1b4f2b8cd21d10fbb0c3b9 : SUCCESS in 23m 34s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c643ff336eb740f4b127dcde99644004 : FAILURE in 16m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed43800c4b3d443f85598d56ad3f492f : SUCCESS in 53m 42s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/82ef3e3ffddf421fa47e2931507de799 : SUCCESS in 43m 22s (non-voting)\n\nWarning:\n  Comments left for invalid file keystone/api/auth.py\n  Comments left for invalid file keystone/tests/unit/test_v3_auth.py","accounts_in_message":[],"_revision_number":1},{"id":"4fdde02a30f6738ba5476bcfdb083265a0513f11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 06:48:48.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dcca92b81a58d07098e27a43c014789b582fcd2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 07:53:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bca5753e4124b13a8bac7bf76066522\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9e1caba63bc24c4a99c082f4b7cfe207 : FAILURE in 4m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5cc7835c97d348858a61bb62b72a7baa : TIMED_OUT in 1h 01m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7d3239636eb87f0c86afbd182425bf618052d948","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 08:33:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b22013059644de3849c48885ba80ad6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02f92a78e93c4e89a6648f228018544e : SUCCESS in 17m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4f0bb4c51874275a9b7cb913819f9c0 : FAILURE in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cebdf7c7a8949168c58215ee9aff111 : SUCCESS in 13m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e06f9c1b152451a870a22c20b7ae675 : SUCCESS in 9m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55d30defdac443fab367089381414f4a : FAILURE in 15m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ae744c0ca244fba8d06a61d76467c6d : SUCCESS in 11m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/72b33cf9729f4563896e757877c24d7a : SUCCESS in 59m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef4b2af6675d4c4098b529fa7e3dda96 : SUCCESS in 1h 28m 57s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/709d2ad5751c41dda0362b39b02c93a7 : SUCCESS in 31m 24s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/3a5a7d71f57b4bec90048868d2bfc55f : SUCCESS in 29m 31s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c317a3b4d4647228b72c89343ccf2bc : FAILURE in 26m 10s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/0a08b03474724646a0975b4c90013f37 : SUCCESS in 27m 55s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c464f0aac87747a492ec2c24a4712750 : SUCCESS in 17m 05s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/7f2e6193ae3a42dcb4a6188d565e4868 : FAILURE in 17m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d35ba443f7974545aa8a3687e88c4f6a : SUCCESS in 58m 40s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/d8e4546f7e31447f9a876d8b1a38fcbe : SUCCESS in 38m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d3610845a70ffcf6529b7974153795332b6c577","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 10:41:24.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e268682159f1179755d36227089affc52bb16e8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 10:46:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d4a05b6943c4315a13a7f982d9f564e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3e300013224f4e9b93eeb2701fd96cac : FAILURE in 4m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b75f2f57fae445ceaf8a62d36b119a56 : FAILURE in 3m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6ee142c5c5a1bb8f80386b1b6f48b33fa046c70d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 12:28:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be52495192c146a1b3d71845839e1e68\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/722d2d22d43e4489bca7493e5fa16869 : SUCCESS in 14m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4125ede99e6544fa841ba07900dc5eca : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a491b7a02a534bf1a4de3b70185544ad : SUCCESS in 14m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3138e57686a34f148cd51919b14f6bc8 : SUCCESS in 13m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/df3872dc84d94215933c247775961059 : FAILURE in 17m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59d76a5ed1384ca6820ce1522a5964a9 : SUCCESS in 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/2f43b0902d71438e9bdf2217ffbf220f : SUCCESS in 54m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c681f31eeaec464db3a7c865e1db0a7f : SUCCESS in 1h 40m 17s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a2cb15ae4f1b419faf3d2ebd8f0593fe : SUCCESS in 30m 39s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/329266bb10db487abdb20c8f9c6de1c8 : SUCCESS in 34m 20s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d85701681b34425fa277295de4c88900 : FAILURE in 19m 37s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/96c0c881e94e46f98445038354557e40 : SUCCESS in 33m 11s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/a3a976fe83ac4fb98b8a67dff5e28abf : SUCCESS in 22m 01s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/fb857e53c6f34fc7950c122c63878d96 : FAILURE in 21m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/733fb4c659084f1cb4facfed66eece0c : SUCCESS in 47m 27s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/04c9e3b5e3fb47d5b7a3a83cb847a62e : FAILURE in 34m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3d6e05e1a8ff40e8bd281e4308e15807d82fc595","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 18:36:13.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"26d2c21cc263ef1e240af15cbe41fa129c2bd9da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 19:40:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/400c2999903b48e690635db90d3c3ac2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f16de2b406a34897b665ea570ed73bad : FAILURE in 4m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f679d29c89034bc99ecde60024f174de : TIMED_OUT in 1h 01m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"051a20d678d5a5f249d185ce683f56ba64c92caf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 20:18:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8a225298eae434088128c7ec8d932ee\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db34e2ae2db940dbb21102c1e03a65b1 : SUCCESS in 17m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f57eea74def0493985be9ba0f16cbbb9 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/191ce62f264b49cdb90782fd2baf5846 : SUCCESS in 14m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a11a40138a6748fe8fadba2fc6098553 : SUCCESS in 13m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca0290ebbd46455a8e1a0165dd2b8f0d : FAILURE in 16m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfdab9025c6b40de8d2c9accef9e483b : SUCCESS in 13m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/ed41cbd8a6a74a3f898fc5dc139ee200 : SUCCESS in 1h 02m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/033ef32c06514ce099c69e82d8049cb2 : SUCCESS in 1h 40m 01s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/48fa55c4b2214e149109cf0732631db0 : SUCCESS in 31m 16s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/2a78849ee69a4d318741874cd42387b8 : SUCCESS in 36m 50s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e9119220409451b9e3d07b029b8644d : FAILURE in 18m 28s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/fbd6b01d68434abd8af2e06977114f3a : SUCCESS in 24m 27s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/4f727553fdf64784bd41b175d756f3e6 : SUCCESS in 21m 12s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ce7e00aa283045999e47c7f93a9aab67 : FAILURE in 26m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb729e3723bd4ebf89e18f5061bb79f4 : SUCCESS in 44m 49s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/01a7e9e9c7aa49859cbd5482122d47fb : FAILURE in 32m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6a9f2f48a94e623dd6aee4bf77e50916b1d428dc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-26 07:56:25.000000000","message":"Abandoned\n\nLanded in other change","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ae92ad33d03f8196729c9e2dfbcf9072a2646ba9","revisions":{"a91a4de12087f221e0dfccb74b8d5ce7a9eecc5a":{"kind":"REWORK","_number":1,"created":"2024-07-11 19:44:00.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/96/923996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/96/923996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/96/923996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/96/923996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/96/923996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/96/923996/1"}}},"commit":{"parents":[{"commit":"270849e40c74e5ea582138e1d5370f4b7bead07a","subject":"Blackify keystone.application_credential","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/270849e40c74e5ea582138e1d5370f4b7bead07a"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:11:12.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:11:12.000000000","tz":120},"subject":"Blackify keystone.application_catalog","message":"Blackify keystone.application_catalog\n\nChange-Id: I7eca91e5775919cfd78b9825cdd7a9808037b1aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a91a4de12087f221e0dfccb74b8d5ce7a9eecc5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a91a4de12087f221e0dfccb74b8d5ce7a9eecc5a"}]},"branch":"refs/heads/master"},"4f110a1dd29661585abded791b293b7f0c00f107":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-12 06:48:48.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/96/923996/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/96/923996/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/96/923996/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/96/923996/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/96/923996/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/96/923996/2"}}},"commit":{"parents":[{"commit":"e7b81c551583d875af20eb8db07dacf6c121c7b2","subject":"Blackify keystone.application_credential","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e7b81c551583d875af20eb8db07dacf6c121c7b2"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:11:12.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 06:44:02.000000000","tz":120},"subject":"Blackify keystone.application_catalog","message":"Blackify keystone.application_catalog\n\nChange-Id: I7eca91e5775919cfd78b9825cdd7a9808037b1aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4f110a1dd29661585abded791b293b7f0c00f107"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4f110a1dd29661585abded791b293b7f0c00f107"}]},"branch":"refs/heads/master"},"f618361a24b46a632355e6ff7ee8851ee28c05ad":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-07-12 10:41:24.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/96/923996/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/96/923996/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/96/923996/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/96/923996/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/96/923996/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/96/923996/3"}}},"commit":{"parents":[{"commit":"24556745482c20aefaa65339474e9328439bc0cb","subject":"Blackify keystone.application_credential","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/24556745482c20aefaa65339474e9328439bc0cb"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:11:12.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 10:41:24.000000000","tz":0},"subject":"Blackify keystone.application_catalog","message":"Blackify keystone.application_catalog\n\nChange-Id: I7eca91e5775919cfd78b9825cdd7a9808037b1aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f618361a24b46a632355e6ff7ee8851ee28c05ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f618361a24b46a632355e6ff7ee8851ee28c05ad"}]},"branch":"refs/heads/master","description":"Rebase"},"ae92ad33d03f8196729c9e2dfbcf9072a2646ba9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-07-12 18:36:13.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/96/923996/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/96/923996/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/96/923996/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/96/923996/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/96/923996/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/96/923996/4"}}},"commit":{"parents":[{"commit":"d5db9992d153214ac94a13f26a03b048d635444c","subject":"Blackify keystone.application_credential","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d5db9992d153214ac94a13f26a03b048d635444c"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:11:12.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 18:30:33.000000000","tz":120},"subject":"Blackify keystone.application_catalog","message":"Blackify keystone.application_catalog\n\nChange-Id: I7eca91e5775919cfd78b9825cdd7a9808037b1aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ae92ad33d03f8196729c9e2dfbcf9072a2646ba9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ae92ad33d03f8196729c9e2dfbcf9072a2646ba9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
