)]}'
{"id":"openstack%2Fkeystone~923998","triplet_id":"openstack%2Fkeystone~master~Idea3bedd6e5c2261679c995c566e8f23d9f387a1","project":"openstack/keystone","branch":"master","topic":"renovate","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-07-26 07:56:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idea3bedd6e5c2261679c995c566e8f23d9f387a1","subject":"Blackify keystone.oauth","status":"ABANDONED","created":"2024-07-11 19:44:00.000000000","updated":"2024-07-26 07:56:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c0ccb67f950be7fc3d21457f141da183ed20cf95","_number":923998,"virtual_id_number":923998,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-12 20:12:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-11 20:47:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-11 21:28:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"80c18e33a6f80dcbe68d3fe9aafe222afc8d0981","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-11 19:44:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"611c498168736064ecabd7c6bb6b915e70defadb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 20:47:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91e56cf9c535487a8b844f86381b9bc5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1c1dff5501f3484fa747880cd183cacf : TIMED_OUT in 1h 01m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c4b6c03c771645138c1e4b17f99b5de9 : FAILURE in 3m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c00b56dad4d69e6865baa0270b25ff115bc6a27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 21:28:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75729fc3929748b0b6afa3ca4b472cd1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/486f41edf56f4daabf9f916575b7fbca : SUCCESS in 13m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61f756c224304b98b5839eab3853e4c0 : FAILURE in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de57fde44f8546008afb299644568aed : SUCCESS in 12m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eae851b3ab3c4a048cc960adb3b68698 : SUCCESS in 11m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c21af182512f4c5ea0085068567d1fcd : FAILURE in 17m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d2813f1e89f4260862d0a8542f9e07d : SUCCESS in 13m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/2182f592d4df41799e3bce08a9443271 : SUCCESS in 1h 03m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/940bed8772454174af98a88db2687ca1 : SUCCESS in 1h 24m 48s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/acb87d5fe2b74ff481ab6c5b74638521 : SUCCESS in 26m 07s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/20f615e875ec4deb8924be1bb2b7feee : SUCCESS in 46m 36s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42301ca5fe3d468683ff377be1771d04 : FAILURE in 28m 32s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/d475c4ce5f90445f9d5893f7d240a11b : SUCCESS in 25m 20s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/2e4c77ff7aae458d9702fe97e1dde22e : SUCCESS in 26m 35s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/90228685693145c1a19ce7507fb5669a : FAILURE in 22m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47590c2d7ee34a978a5d2d6aecfbdfd7 : SUCCESS in 42m 02s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/f21bf5ce67444b6091a0b651463c27d4 : SUCCESS in 41m 40s (non-voting)\n\nWarning:\n  Comments left for invalid file keystone/api/auth.py\n  Comments left for invalid file keystone/api/users.py\n  Comments left for invalid file keystone/identity/backends/ldap/common.py\n  Comments left for invalid file keystone/identity/core.py\n  Comments left for invalid file keystone/tests/unit/test_v3_auth.py","accounts_in_message":[],"_revision_number":1},{"id":"85d5cdf95aae8fb84d82bd60a6554fc74f988eff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 06:48:48.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f06f03b7f5a6c3e943c91ab544b0febbbe06704b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 07:01:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6e0e5b3b30c49ef8631b63c24c92059\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7c952a4b48fe4c608d552d68e89ddf83 : FAILURE in 4m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/57295294645b4a8c922be43f736aab41 : FAILURE in 3m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"801982a19d9801a27df7f14888d39765ca92fb64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 08:01:41.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"c1f2d755db24bb51fed45f2056616b04a9174163","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 08:10:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20bbaa3dffd3402ea68fcdf28538513b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/20480f013bee45d6937c5d8e1ea3d856 : FAILURE in 4m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4bc10040a14e470586b91343e272632c : FAILURE in 4m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7e29f2cd169e2792c0bb3fce89a6e5ae2c498d4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 09:21:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45a46818b2ff4e7b823b1a122258bc0d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/963fdf7293044edbb8c40ce7ccfa5848 : SUCCESS in 13m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4d9933bad2e4b83b0914f1448039b8c : FAILURE in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96b0d49db71943d1a66d116984ffe936 : SUCCESS in 13m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c02b69bfe4c64f3e9c65b595d765cbca : SUCCESS in 9m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/92ae6a50a3414f31b33e78fdd84c5322 : FAILURE in 13m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eba023cd49b49d0839212aa5b1212d1 : SUCCESS in 13m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/d9cc7afc6b63496b87f8aefd753d0534 : SUCCESS in 1h 08m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/525b85b98baf498b93a76d449f6f1c38 : SUCCESS in 1h 12m 52s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/462eb5ada3444dbbaaaec9224b401771 : SUCCESS in 19m 47s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/e29b0f528c1c44f08a08621a924fdd47 : SUCCESS in 34m 13s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8039d2bb158a4ef9a867cdf998f8eea3 : FAILURE in 16m 44s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/ec4639c162c24dca9ae34109e857e789 : SUCCESS in 20m 21s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/74fb661d777c459594f00ae42d59c421 : SUCCESS in 12m 27s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/d4796ccac01448178bb449271631bdae : FAILURE in 13m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61ab3e5a238943469d531329e814a4b1 : SUCCESS in 47m 48s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/a0038ff8a7e0430aaea3f7d095608431 : SUCCESS in 36m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6a706973be0051eb2cb4329728b38edc7fc23e7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 10:41:37.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7547f24267c1e3500db379b748050bc76c4ffe57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 10:56:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5296e3a2fef45a2b7eed1a60e814340\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad2526b2c4b04ce9a152d57a26b18ee4 : RETRY_LIMIT in 2m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/34029a6374954f608532748a47373e86 : FAILURE in 3m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"89bd424db4ff29402c80ab7574fa1d97d061940e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 12:09:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcf733450fd3463394dea8b498c9fcf8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9355cbedc73490e88f2ab97a24692bc : SUCCESS in 18m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fed18685a94e49f284ffd8add19ceefe : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae4e65c0a1974a3b915629902d84d2af : SUCCESS in 12m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f954d0ed1894a3283f4e6f848690bd7 : SUCCESS in 11m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/579f23d73bc54035ac20333eba2235bc : FAILURE in 17m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d425a3dff0e6444c8a7890e5be9ede4f : SUCCESS in 13m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/0cdbc30860d2443297687c8ff9e262de : SUCCESS in 52m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da47635d9d34442989628c3b0f6d185d : SUCCESS in 1h 26m 29s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/34cdf567e2ac4617ad96c52e2f1fb05d : SUCCESS in 29m 33s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/43a15fdc47d640eb886f5067bc4758fe : SUCCESS in 42m 11s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fcc89fab16e548288f6dab40a0a48b3d : FAILURE in 25m 04s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/3b3d116e5d4b4cfb832653d754877fee : SUCCESS in 24m 16s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e4b3de4486f14223ae8b419d77058138 : FAILURE in 13m 38s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/1d6d02c57e0b4bd69bdd5f43a8b56333 : FAILURE in 15m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ec9d5e3d78d4003a95150a914df76de : SUCCESS in 58m 45s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/faab98bfb0e44fb199c0396261866a34 : SUCCESS in 43m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c47ed0824082a983dbec0f51df32e6b1db52a95b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 18:36:13.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"6abaa2113bec6600f5760938773080ff7e615d59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 19:25:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0eaec95a203a46ab82fbd76c63a1f0c3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15db664cbba546868e82c74988f0f0e3 : FAILURE in 5m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/50d2b387691043a39dbdf9098597fbe7 : SUCCESS in 46m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"079886d018a9d83d445a2ae8ef1137aac0cf52bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 20:12:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4234d27c141d42be88eb81c9887a879d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce0cdb1930b44efe8055602047c3b4c5 : SUCCESS in 18m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2e775cd525649bcab9b783c3725ff95 : SUCCESS in 7m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44758bf91091460aa34207ba26518fde : SUCCESS in 12m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec9cb45d8c1e4518848a37e9660dcb1e : SUCCESS in 10m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/19712999d8af4c41be4121973ca9e302 : FAILURE in 13m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f384e907bd84f4ca428c2540c011fbd : SUCCESS in 12m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/dff2ff12973a43c3bff11f1cdcca43e1 : SUCCESS in 1h 04m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2af0c3c78bff47628010ff8c1c86ef83 : SUCCESS in 1h 30m 51s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/db87e0191ff64774b0b006fbb1cff2f9 : SUCCESS in 26m 26s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/fd2c89f2d6414ff1aec0179c9e37b147 : SUCCESS in 44m 19s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3776c286294b4b908a97f6aa8d94ff9c : FAILURE in 25m 09s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/ccfbcf0781954144bece94d0e3dc06fc : SUCCESS in 27m 35s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/adde68e2f200404c9201370ff8e65b1d : SUCCESS in 10m 03s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/d65c130a16514010b33a91177b9030da : FAILURE in 18m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eda3c5fef62f4c0b86e61f201062a4ee : SUCCESS in 52m 16s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/14a3edb2ea554102a455516b778d1029 : SUCCESS in 40m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c0ccb67f950be7fc3d21457f141da183ed20cf95","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-26 07:56:01.000000000","message":"Abandoned\n\nLanded in other change","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6c249a177c5e901a91cc426a6d923b4113480192","revisions":{"6fd2f0148c45d9111dd5e4df655d6124e0eaecb3":{"kind":"REWORK","_number":1,"created":"2024-07-11 19:44:00.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/98/923998/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/98/923998/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/98/923998/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/98/923998/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/98/923998/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/98/923998/1"}}},"commit":{"parents":[{"commit":"b98efd611cc7a1598478996d2aa647a2fa618ad6","subject":"Blackify keystone.identity","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b98efd611cc7a1598478996d2aa647a2fa618ad6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:13:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:13:57.000000000","tz":120},"subject":"Blackify keystone.oauth","message":"Blackify keystone.oauth\n\nChange-Id: Idea3bedd6e5c2261679c995c566e8f23d9f387a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6fd2f0148c45d9111dd5e4df655d6124e0eaecb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6fd2f0148c45d9111dd5e4df655d6124e0eaecb3"}]},"branch":"refs/heads/master"},"5e18eb0a013daf7ca2b1bf04a56a41237704ff57":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-12 06:48:48.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/98/923998/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/98/923998/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/98/923998/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/98/923998/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/98/923998/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/98/923998/2"}}},"commit":{"parents":[{"commit":"1a5575d4ae2fc174d7e6ea54e933d98b0f02c673","subject":"Blackify keystone.identity","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1a5575d4ae2fc174d7e6ea54e933d98b0f02c673"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:13:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 06:44:16.000000000","tz":120},"subject":"Blackify keystone.oauth","message":"Blackify keystone.oauth\n\nChange-Id: Idea3bedd6e5c2261679c995c566e8f23d9f387a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5e18eb0a013daf7ca2b1bf04a56a41237704ff57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5e18eb0a013daf7ca2b1bf04a56a41237704ff57"}]},"branch":"refs/heads/master"},"1d15e668733e920f7adc0779d0677c03787990b8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-07-12 08:01:41.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/98/923998/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/98/923998/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/98/923998/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/98/923998/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/98/923998/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/98/923998/3"}}},"commit":{"parents":[{"commit":"9a5b58232bd77c1c186dda40fd39d6aadb9f8e0a","subject":"Blackify keystone.identity","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9a5b58232bd77c1c186dda40fd39d6aadb9f8e0a"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:13:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 07:12:38.000000000","tz":120},"subject":"Blackify keystone.oauth","message":"Blackify keystone.oauth\n\nChange-Id: Idea3bedd6e5c2261679c995c566e8f23d9f387a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1d15e668733e920f7adc0779d0677c03787990b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1d15e668733e920f7adc0779d0677c03787990b8"}]},"branch":"refs/heads/master"},"c0023729cd1ccab1bf7feddaa19a42e423385486":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-07-12 10:41:37.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/98/923998/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/98/923998/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/98/923998/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/98/923998/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/98/923998/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/98/923998/4"}}},"commit":{"parents":[{"commit":"0ebac26aa042d2f0fce61d665f5b903a243b1b2a","subject":"Blackify keystone.identity","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0ebac26aa042d2f0fce61d665f5b903a243b1b2a"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:13:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 10:41:37.000000000","tz":0},"subject":"Blackify keystone.oauth","message":"Blackify keystone.oauth\n\nChange-Id: Idea3bedd6e5c2261679c995c566e8f23d9f387a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c0023729cd1ccab1bf7feddaa19a42e423385486"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c0023729cd1ccab1bf7feddaa19a42e423385486"}]},"branch":"refs/heads/master","description":"Rebase"},"6c249a177c5e901a91cc426a6d923b4113480192":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-07-12 18:36:13.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/98/923998/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/98/923998/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/98/923998/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/98/923998/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/98/923998/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/98/923998/5"}}},"commit":{"parents":[{"commit":"74a9eef6fb3c6dd85cbc9d3397cfa295a381d203","subject":"Blackify keystone.identity","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/74a9eef6fb3c6dd85cbc9d3397cfa295a381d203"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:13:57.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 18:33:29.000000000","tz":120},"subject":"Blackify keystone.oauth","message":"Blackify keystone.oauth\n\nChange-Id: Idea3bedd6e5c2261679c995c566e8f23d9f387a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6c249a177c5e901a91cc426a6d923b4113480192"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6c249a177c5e901a91cc426a6d923b4113480192"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
