)]}'
{"id":"openstack%2Fkeystone~923999","triplet_id":"openstack%2Fkeystone~master~I18bcda7001783cb835f552e3868b696b5187b8b5","project":"openstack/keystone","branch":"master","topic":"renovate","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-07-26 07:55:49.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I18bcda7001783cb835f552e3868b696b5187b8b5","subject":"Blackify keystone.federation","status":"ABANDONED","created":"2024-07-11 19:44:00.000000000","updated":"2024-07-26 07:55:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a983221a527f8d1761a8ed99651fbfc0b44f754c","_number":923999,"virtual_id_number":923999,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-12 20:11:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-11 20:48:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-11 21:36:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba9710f32a3ec082172bff91523d679483270c2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-11 19:44:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b92466450aa0ed24326b8183eb4bdb362c669759","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 20:48:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3e48c0449c5487fb3ab3e03997bf76d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/61ae66c26e6a40b98f3dc57ed0034516 : TIMED_OUT in 1h 01m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/665e03bfd80149a8a6dfdd71b8e24a89 : FAILURE in 3m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f32ac350f81ef4b5dd718cebbf9371ba34d0822b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 21:36:57.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f3344675cd74d74b9e9bb550545bc9e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/097225da2e2b4aba8adc5a6c48bed16e : SUCCESS in 15m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d33975375ba48aba6dbe2fd0dc5df99 : FAILURE in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da3125beab304ca8b365bb7cf156eabc : SUCCESS in 12m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c598534e09d43b587abf2873ee7a82a : SUCCESS in 14m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0f5a9603e42d4d439dbce2c0542e27f5 : FAILURE in 15m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3878c0fa21f462fa74a57a491079dc4 : SUCCESS in 12m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/fa97e75b68234db8ba162af3553f8710 : SUCCESS in 1h 01m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d317f1f8ff04daf91f9a7751362327b : SUCCESS in 1h 45m 12s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/9935d8779c6a4c22b2e7bf9bd22ff408 : SUCCESS in 29m 17s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/b8a1443e27ac40d08f4a3bce09dbcb02 : SUCCESS in 38m 01s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20e81ba77dd445a0b191810f3fe1a3ec : FAILURE in 24m 27s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/5910c1454db54bb0acc2524e22e7bf0b : SUCCESS in 32m 55s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/1b1c944e0b1f42d1af95a694c97519a8 : SUCCESS in 23m 12s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/dda3b62b174a468a88877892f069aada : FAILURE in 22m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea08e41a7efb4d8d99ea57450e06f6d6 : SUCCESS in 1h 00m 49s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/d7868ebd617d43aba179479562171dd6 : SUCCESS in 39m 13s (non-voting)\n\nWarning:\n  Comments left for invalid file keystone/api/auth.py\n  Comments left for invalid file keystone/api/users.py\n  Comments left for invalid file keystone/identity/backends/ldap/common.py\n  Comments left for invalid file keystone/identity/core.py\n  Comments left for invalid file keystone/tests/unit/test_v3_auth.py","accounts_in_message":[],"_revision_number":1},{"id":"9163375e174c92558fe04bfd4a596bead65bad56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 06:48:48.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1832d8fb8d20af80c54805a61abe8b048e07d2c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 07:53:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1a0b64fd72740608c65f885043f835f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/93de54006d9c497fada4a3f62760db6f : TIMED_OUT in 1h 01m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/12e902a1897e4337b1e21ab79a807d7a : FAILURE in 4m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ee8183634e7c91ca4d3ccafbccda2035ee7cf4a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 08:01:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2a52a932408fcd0ba73ca2416bb2ded6ee83cf9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 08:15:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ebf905e01a7490197ff1182d324702b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9bb882dcb74e4159a80866de0c05139a : FAILURE in 5m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c255cd3118504d1fa51900cf3c133fe6 : FAILURE in 4m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e0ca8f57dc5f26e353fd6727ed9ec1177067b046","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 09:27:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9863b2e95c6448a8b789a16795689dbc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c8668d149e02495e9f0c2e6086190577 : SUCCESS in 5m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab88aa53dce44cd3bdcf43f7697c04dc : SUCCESS in 14m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/838a5b79be574793bc67f09a85e7d5d1 : FAILURE in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd59759d89b14d4ab057e75750a203b5 : SUCCESS in 12m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85c760cade154943b3fe4d4e02f2922c : SUCCESS in 12m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2eb979ee72f54f66891dc8b4d443fe35 : FAILURE in 12m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb45c3584d7847e598a1f6c304182afb : SUCCESS in 11m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/e75060362df04b11b0a219ae994d886d : SUCCESS in 1h 01m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd68200b4b7c450683607caed43334c8 : SUCCESS in 1h 20m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a6096c2fd2647e7b5c14909f8600c46 : SUCCESS in 13m 25s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/240284ac3d454479a36f9ba0d1ee7aaa : SUCCESS in 30m 00s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/5dc4d633205743ad80a184a6041ee9f8 : SUCCESS in 33m 27s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3215ee71929e48fd8b6197d8d9d3e1f7 : FAILURE in 27m 54s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/2b9bee22d272487490db34a9151792e2 : SUCCESS in 29m 37s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/58177188481d4ea1acf0c3e0ba78111b : SUCCESS in 18m 38s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/502b0cad0cf44c9e9e5d32532244ce69 : FAILURE in 25m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b339e00734d4e0aa0920ce5433dca14 : SUCCESS in 55m 08s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/6ce36415711041e4acdae6b2d99161b6 : FAILURE in 40m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"920fad786ddc9abcede0d454352c86ff86367e5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 10:41:43.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"4791f42abbd74dff8e1641d50e492f4a7f8c4ded","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 11:03:34.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b29927cbea2b45099728f25f3aa31e8d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/241857e150984ee38a7e55e34ad32c33 : SUCCESS in 15m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4a0b4aea4f6a43dca3f318502ab36411 : SUCCESS in 16m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c343a9b4a2ef441f222b902082dfaf603f4f56a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 12:06:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17ca8f533ae343b5bd4fed4966f6f48f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c40c3863d2d44c39a385ceb72c922853 : SUCCESS in 5m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a25d29be4bc4d7bb144e872aaf072a4 : SUCCESS in 15m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf40c066c07741ba8072e8bf498ec761 : SUCCESS in 7m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e21cb07a750242f8836ebd4c535be751 : SUCCESS in 11m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4546c078fb7c4749a89f91f7fbb0770f : SUCCESS in 10m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/01f968909ced490cb95f5b7ae5f00c9a : FAILURE in 16m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e756bca4a424e7bb4db8fbc838e8724 : SUCCESS in 12m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/fcbbc58ad66142a48d3c0643c0559287 : SUCCESS in 1h 08m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8acb213a5ae4aeba5efa9724e2992ac : SUCCESS in 1h 20m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d72ec93e72b42f583fa5c60c6b9adaf : SUCCESS in 11m 42s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/32008e3877064909855b74b57d9a7b72 : SUCCESS in 28m 14s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/627b67587bef46049c38e8e887c388d5 : SUCCESS in 38m 26s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f9841b1beb74764925978b532171646 : FAILURE in 25m 45s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/fe9a6d3882f541c4a2b5b1209a1ce185 : SUCCESS in 23m 55s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/723c59037b4b4e9e9596aad91bb6e4d3 : SUCCESS in 12m 11s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/3dca92b89ebe4dc59596c4d1e7302c7d : FAILURE in 26m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c5fa1b82fce4c42b8e5c180355646db : SUCCESS in 59m 23s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/51f2d5ed9fd04b0499284c8ec20f5459 : SUCCESS in 38m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7320ff1d88a18319ae6c43e99d7bd5c157c06903","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 18:36:13.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"b9cac6f91c506ca8b7b531a81a932adf75ef300c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 19:21:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d3367e8ce484386a6237c7bdf44cc3f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/613f09dff61a4d3ebca784f4cd91ef01 : FAILURE in 4m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a30073ff0ea14c38bbdb71d00d6c92bb : SUCCESS in 41m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2c27c00aa01ab93a29ac7be6b3ca0b95f9739f7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 20:11:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/700673a6349242d09e64e79fd4ea2cfd\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/508b160e56f4488cb9e470966b486b13 : SUCCESS in 6m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37be231ae7c94eb5b05174b469877cc7 : SUCCESS in 18m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4def04b501e44010877540da518861ae : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f92a3878215849299530999cbd23757f : SUCCESS in 13m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb3ffb95f91c4f4d91c762040330c12e : SUCCESS in 11m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b512c0d9a807463c98601b298799a173 : FAILURE in 12m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85d64e07e0674949a16d34dd8eb26b2b : SUCCESS in 11m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/9213fa73a63446d2b0f8b248251aa949 : SUCCESS in 58m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/037deda463704361b78eee04345f3a01 : SUCCESS in 1h 32m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f2910574ed440ef98483b19030047a7 : SUCCESS in 10m 25s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/4f4f494fe3f94af3a0b0ece35602c414 : SUCCESS in 27m 13s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/889e4ac39ad2414b97869e0af7504313 : SUCCESS in 43m 10s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6910be39557249b68ba3b33fe0d3294d : FAILURE in 22m 50s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/dd821c5f0d1249b893e2d6a9f6a33b93 : SUCCESS in 22m 37s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c24cdc12a44945efbfbd852f1a3cf6c6 : SUCCESS in 20m 15s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ab59c3a5485349d5bb252ed2c4d07857 : FAILURE in 21m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e646a36e7cf415bb3c82eafb362208c : SUCCESS in 53m 58s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/654e2d18ad1f45219d2297689e4b4205 : FAILURE in 40m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a983221a527f8d1761a8ed99651fbfc0b44f754c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-26 07:55:49.000000000","message":"Abandoned\n\nLanded in other change","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d26dee47c9c39d33e7470c64b3cdc0ecd5f0778e","revisions":{"ea1d8a2cef317ce8d20162d3dd03312f824d7469":{"kind":"REWORK","_number":1,"created":"2024-07-11 19:44:00.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/923999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/99/923999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/923999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/923999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/923999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/99/923999/1"}}},"commit":{"parents":[{"commit":"6fd2f0148c45d9111dd5e4df655d6124e0eaecb3","subject":"Blackify keystone.oauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6fd2f0148c45d9111dd5e4df655d6124e0eaecb3"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:15:39.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:15:39.000000000","tz":120},"subject":"Blackify keystone.federation","message":"Blackify keystone.federation\n\nChange-Id: I18bcda7001783cb835f552e3868b696b5187b8b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ea1d8a2cef317ce8d20162d3dd03312f824d7469"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ea1d8a2cef317ce8d20162d3dd03312f824d7469"}]},"branch":"refs/heads/master"},"68c43c30d6f89d9a6546a3b23c63b4887b3cfc88":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-12 06:48:48.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/923999/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/99/923999/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/923999/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/923999/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/923999/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/99/923999/2"}}},"commit":{"parents":[{"commit":"5e18eb0a013daf7ca2b1bf04a56a41237704ff57","subject":"Blackify keystone.oauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5e18eb0a013daf7ca2b1bf04a56a41237704ff57"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:15:39.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 06:44:22.000000000","tz":120},"subject":"Blackify keystone.federation","message":"Blackify keystone.federation\n\nChange-Id: I18bcda7001783cb835f552e3868b696b5187b8b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/68c43c30d6f89d9a6546a3b23c63b4887b3cfc88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/68c43c30d6f89d9a6546a3b23c63b4887b3cfc88"}]},"branch":"refs/heads/master"},"a7d42ed456bb1e62916eefd18919604dcbe38115":{"kind":"REWORK","_number":3,"created":"2024-07-12 08:01:41.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/923999/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/99/923999/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/923999/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/923999/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/923999/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/99/923999/3"}}},"commit":{"parents":[{"commit":"1d15e668733e920f7adc0779d0677c03787990b8","subject":"Blackify keystone.oauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1d15e668733e920f7adc0779d0677c03787990b8"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:15:39.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 07:16:52.000000000","tz":120},"subject":"Blackify keystone.federation","message":"Blackify keystone.federation\n\nChange-Id: I18bcda7001783cb835f552e3868b696b5187b8b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a7d42ed456bb1e62916eefd18919604dcbe38115"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a7d42ed456bb1e62916eefd18919604dcbe38115"}]},"branch":"refs/heads/master"},"5f93d5588a0c40af1428d3169e8eb30bf10defc5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-07-12 10:41:43.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/923999/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/99/923999/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/923999/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/923999/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/923999/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/99/923999/4"}}},"commit":{"parents":[{"commit":"c0023729cd1ccab1bf7feddaa19a42e423385486","subject":"Blackify keystone.oauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c0023729cd1ccab1bf7feddaa19a42e423385486"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:15:39.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 10:41:43.000000000","tz":0},"subject":"Blackify keystone.federation","message":"Blackify keystone.federation\n\nChange-Id: I18bcda7001783cb835f552e3868b696b5187b8b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5f93d5588a0c40af1428d3169e8eb30bf10defc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5f93d5588a0c40af1428d3169e8eb30bf10defc5"}]},"branch":"refs/heads/master","description":"Rebase"},"d26dee47c9c39d33e7470c64b3cdc0ecd5f0778e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-07-12 18:36:13.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/99/923999/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/99/923999/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/923999/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/923999/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/923999/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/99/923999/5"}}},"commit":{"parents":[{"commit":"6c249a177c5e901a91cc426a6d923b4113480192","subject":"Blackify keystone.oauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6c249a177c5e901a91cc426a6d923b4113480192"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:15:39.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 18:33:29.000000000","tz":120},"subject":"Blackify keystone.federation","message":"Blackify keystone.federation\n\nChange-Id: I18bcda7001783cb835f552e3868b696b5187b8b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d26dee47c9c39d33e7470c64b3cdc0ecd5f0778e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d26dee47c9c39d33e7470c64b3cdc0ecd5f0778e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
