)]}'
{"id":"openstack%2Fkeystone~924003","triplet_id":"openstack%2Fkeystone~master~Ia5a57323d4f9be9d78d79c1bc33803cbe4fc1f41","project":"openstack/keystone","branch":"master","topic":"renovate","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-07-26 07:54:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia5a57323d4f9be9d78d79c1bc33803cbe4fc1f41","subject":"Blackify keystone.api remains","status":"ABANDONED","created":"2024-07-11 19:44:00.000000000","updated":"2024-07-26 07:54:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f21b19736ab7abd0064ce007aed5d30380e96036","_number":924003,"virtual_id_number":924003,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-12 20:16:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-11 20:48:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-11 21:54:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62dc6bfb0c2e2cdce69f99a7ed947591b7b6c3f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-11 19:44:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50ab5b88a9dbd8e66f25517406fd1925574fc690","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 20:48:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/061c2919a239462ea4724dcf25ddb6c7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c1f798bdfd246cd8c65f87439872750 : TIMED_OUT in 1h 01m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5962f205fbd34f809d2045e1a804ed85 : FAILURE in 3m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d68d3b87426f71614f5aa297efb02f4900858d92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 21:54:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ffc174ac0cb495aa1457b34b1a5d7bd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7284b6f188c0423da2862c32a0097f80 : SUCCESS in 14m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9de9642d35bb4443adadae7d34f9a910 : FAILURE in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c63e32c8bc441bd95caadb788a8dcd1 : SUCCESS in 10m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c9de18b10e6473b80805135ced44868 : SUCCESS in 14m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/87d42b7c50e34975998345dd83bbb1d7 : FAILURE in 15m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd8ce2dc9b6843ab88a957db621fd6bb : SUCCESS in 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/aff3e26d46764db3ac80c791f078f58a : SUCCESS in 1h 17m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e20f76fbf0b74e7791c99e8c0ece2716 : SUCCESS in 1h 39m 45s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d34284a3c41a4e3ca12a7755824c6ed6 : SUCCESS in 27m 47s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/fadaea7a0e2d4baab9e842b4406de967 : SUCCESS in 38m 33s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f58c9fab9e347beacd3dca18acc1bdc : FAILURE in 25m 27s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/3285a9d370594c3c9837649625d573d9 : SUCCESS in 29m 52s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/6da4afade36b49b08df27d69945820b5 : SUCCESS in 18m 06s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c58855c877ef4b52aa3027ed217304e1 : FAILURE in 12m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd136046b7d14e059f3be1cd03871fe1 : SUCCESS in 1h 02m 17s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/02173ba8d25e4b26ab17c01c27ccd379 : SUCCESS in 36m 32s (non-voting)\n\nWarning:\n  Comments left for invalid file keystone/api/auth.py\n  Comments left for invalid file keystone/api/users.py\n  Comments left for invalid file keystone/assignment/core.py\n  Comments left for invalid file keystone/federation/utils.py\n  Comments left for invalid file keystone/identity/backends/ldap/common.py\n  Comments left for invalid file keystone/identity/core.py\n  Comments left for invalid file keystone/resource/core.py\n  Comments left for invalid file keystone/tests/unit/assignment/test_backends.py\n  Comments left for invalid file keystone/tests/unit/test_v3_auth.py","accounts_in_message":[],"_revision_number":1},{"id":"a02f6160ba2d48e71ab5ad7fae1f295b85db0c33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 06:48:48.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e2957d7bd8882a7584cc02aa9fcdc0b973798c40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 07:53:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afeaba9a4a7d48f8875b99128c37c289\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3925da5ad7954c01ad68dbcf4584e765 : TIMED_OUT in 1h 01m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/07d303ea0b9b44b29df797610f03664a : FAILURE in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"42af5948ae67efc1e183abfc19c59921901cdbb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 08:01:41.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"e5e0670c7e904d1a2a7ecbdce551bcad9e5e471a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 09:05:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67757822356d4450ba183c8ae0c9aec5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce41777dd5b1433a9ade81c6c9cda867 : FAILURE in 4m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3a46c66e9fc044ba8fe1081baafc8246 : TIMED_OUT in 1h 01m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e5c184f896dbd49009d063301760b5ba065afd09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 09:50:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33875927ff134d81b0f130864a5fac0a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1601d9dda60748c8af651a21939c8020 : SUCCESS in 13m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81e80c3edc1544f4adfbb1107662d8ff : FAILURE in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/651335b8b21c499aa191304eefad6137 : SUCCESS in 13m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/62877c6a9725463a98c5b66670655277 : SUCCESS in 11m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/46b0dca9f1064aa3a0fe952e07faf173 : FAILURE in 14m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6726edd84ca48ceb5c7bb7dd8667a17 : SUCCESS in 13m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/b175e03ddf7f4cffbd6c7e4618a3c1a2 : SUCCESS in 1h 07m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/393db27e95bf4bdd9c82ae91bd6eb206 : SUCCESS in 1h 40m 24s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/69fac088a45045ba828628570bdf7bd4 : SUCCESS in 23m 47s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/85cc5550d02a4097819d723e99133fac : SUCCESS in 43m 25s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b274c9d4ea634355a29082576d637032 : FAILURE in 29m 00s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/16bff834b7364d30b5f8c6ad0e02a5aa : SUCCESS in 22m 47s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/14edb2be5d7e4071ab5701c020b3b299 : SUCCESS in 22m 27s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/76b86eb9267c452da4aa90eac8203c3e : FAILURE in 21m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08c81f8f1a504f42b18fdb45c9a38cc7 : SUCCESS in 57m 18s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/48d37a3e0a8440828ffa0e7b333495e3 : SUCCESS in 32m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"780fd3ca77f0aba61d295d2ab83dd898b83c0f0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 10:42:08.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a74d4ac4d3500ecae127ae45475eb24730fb428f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 11:01:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80c11719fb924a39ae54caeecfcbebcd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/37ff1e1c72384e64ba738bd98415278a : SUCCESS in 13m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4b58cf222534442a86069628b0e487c3 : FAILURE in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f7ff29d35bee32692454c47640042a3c593d5125","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 12:19:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52add08fd30546f68e76df3d4c2acd34\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f46b54cbb8d6488ebe52a40239eab4f3 : SUCCESS in 12m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3cbf952c076458e9a2f9393ec32f776 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b3e05bad06a48eca91084e89ac6898b : SUCCESS in 11m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85a04670843942dfb97bc7362d346ddd : SUCCESS in 13m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ed5a8e5b5234472bbb3c719e192b49f : FAILURE in 15m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02cf900616304acbadf22aad0871cdc8 : SUCCESS in 13m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/a422f93c4c1646c1b7b80de0165c78e7 : SUCCESS in 48m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91544ab0642449719af489084c0ca752 : SUCCESS in 1h 31m 16s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1e56fcbf80e34426906f13ff38edd46a : SUCCESS in 34m 02s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/f02db256a060457d8c4840e1491fbf03 : SUCCESS in 29m 36s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/981702f6a61647d79807201b724be670 : FAILURE in 26m 33s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/069002be45734d22949d89f467020cc5 : SUCCESS in 25m 37s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/393358f7f80d4cb88ce355d980ae2edd : SUCCESS in 19m 31s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/a568fd2cd44b4c029218683112933ca1 : FAILURE in 19m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b52620fdc02443f186634a223ee7aa19 : SUCCESS in 42m 40s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/7d6d6ccec6fd49a48e1db0e3a49d2c0d : SUCCESS in 46m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c2de2b6c662186a9a614b4edf303116a075f2b98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 18:36:13.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"ef2e16f903e92d5da34c0d6a8f28b74e1d8edd2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 19:40:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3f6c0c569f24a1db119156d811ccc0c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f7e493d96d8041bbbba0be0d847a5a74 : TIMED_OUT in 1h 01m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e24fac5d1e8240c982900fd98641e302 : SUCCESS in 56m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e7fa07798e14a7a2fc08372bd1163915c0fdb119","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 20:16:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae3dc7541d46498da200e043958a9920\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c7e4f8105334181983eed76327fca49 : SUCCESS in 16m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f1d8ccb0f48490fa61d88d95148ec5d : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d17e15fe576b42d3b8ce0f33f6c56bde : SUCCESS in 12m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1d06c8aa7004721b8ff28462c8f6a11 : SUCCESS in 11m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/72422814b71746a49c343e5d2d346f27 : FAILURE in 16m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f50b4a12a4f4968863eebe33d519d9e : SUCCESS in 11m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/1040cf8d1b3a4e77a3737d182dd9c27b : SUCCESS in 1h 08m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe972cf167224a56b2d035ba428c659b : SUCCESS in 1h 32m 08s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/7aeec831c2cb416cae9d73688bb8cb6c : SUCCESS in 31m 49s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/6e3275e738974fe1b8924c6814750e9f : SUCCESS in 38m 58s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2be505bfc9e44284bf144aeb7865fa1e : FAILURE in 25m 55s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/49daabdcbf2e4da29942a1b909491581 : SUCCESS in 33m 10s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/61547100977e471592e65e1cd7d99dbe : SUCCESS in 19m 10s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/d9c25877d78e45b5ac36ebf5dc15c855 : FAILURE in 25m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ac7c32d145a4d4d9eb8b05a92ecb7b0 : SUCCESS in 1h 02m 46s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/37475f8ff5e446b7b75420b254c21238 : SUCCESS in 37m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f21b19736ab7abd0064ce007aed5d30380e96036","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-26 07:54:55.000000000","message":"Abandoned\n\nLanded in other change","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a8be0dd30fdfc90885938df2d2c127d75d885df6","revisions":{"449570b7ccd783c271579fe9a3c8a73a4b5b67bc":{"kind":"REWORK","_number":1,"created":"2024-07-11 19:44:00.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/03/924003/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/03/924003/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/924003/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/924003/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/924003/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/03/924003/1"}}},"commit":{"parents":[{"commit":"86c38a3bc1572a5a5395aafb6312db587e419b1e","subject":"Blackify keystone.limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/86c38a3bc1572a5a5395aafb6312db587e419b1e"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:22:53.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:22:53.000000000","tz":120},"subject":"Blackify keystone.api remains","message":"Blackify keystone.api remains\n\nChange-Id: Ia5a57323d4f9be9d78d79c1bc33803cbe4fc1f41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/449570b7ccd783c271579fe9a3c8a73a4b5b67bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/449570b7ccd783c271579fe9a3c8a73a4b5b67bc"}]},"branch":"refs/heads/master"},"df43ff1f27816518987e9711ec9052eec7f51999":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-12 06:48:48.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/03/924003/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/03/924003/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/924003/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/924003/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/924003/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/03/924003/2"}}},"commit":{"parents":[{"commit":"f78ee97c5626e2898a2c22de2a42bc173c230145","subject":"Blackify keystone.limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f78ee97c5626e2898a2c22de2a42bc173c230145"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:22:53.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 06:44:53.000000000","tz":120},"subject":"Blackify keystone.api remains","message":"Blackify keystone.api remains\n\nChange-Id: Ia5a57323d4f9be9d78d79c1bc33803cbe4fc1f41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/df43ff1f27816518987e9711ec9052eec7f51999"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/df43ff1f27816518987e9711ec9052eec7f51999"}]},"branch":"refs/heads/master"},"a1b1bdda18c1ff6432326acb0fcc5294e0574259":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-07-12 08:01:41.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/03/924003/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/03/924003/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/924003/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/924003/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/924003/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/03/924003/3"}}},"commit":{"parents":[{"commit":"31760c5e6ff42c04575806ad90f979febc0eb5cd","subject":"Blackify keystone.limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/31760c5e6ff42c04575806ad90f979febc0eb5cd"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:22:53.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 07:21:24.000000000","tz":120},"subject":"Blackify keystone.api remains","message":"Blackify keystone.api remains\n\nChange-Id: Ia5a57323d4f9be9d78d79c1bc33803cbe4fc1f41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a1b1bdda18c1ff6432326acb0fcc5294e0574259"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a1b1bdda18c1ff6432326acb0fcc5294e0574259"}]},"branch":"refs/heads/master"},"86af955db2e0f7f1d2593d0fd7b092017fb53066":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-07-12 10:42:08.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/03/924003/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/03/924003/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/924003/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/924003/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/924003/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/03/924003/4"}}},"commit":{"parents":[{"commit":"645de91732ad850c1770a44d80b8e1b1aab02094","subject":"Blackify keystone.limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/645de91732ad850c1770a44d80b8e1b1aab02094"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:22:53.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 10:42:08.000000000","tz":0},"subject":"Blackify keystone.api remains","message":"Blackify keystone.api remains\n\nChange-Id: Ia5a57323d4f9be9d78d79c1bc33803cbe4fc1f41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/86af955db2e0f7f1d2593d0fd7b092017fb53066"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/86af955db2e0f7f1d2593d0fd7b092017fb53066"}]},"branch":"refs/heads/master","description":"Rebase"},"a8be0dd30fdfc90885938df2d2c127d75d885df6":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-07-12 18:36:13.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/03/924003/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/03/924003/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/924003/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/924003/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/924003/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/03/924003/5"}}},"commit":{"parents":[{"commit":"46b37afb3274b2fa75d8c8690b2458b01ac4764b","subject":"Blackify keystone.limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/46b37afb3274b2fa75d8c8690b2458b01ac4764b"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:22:53.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 18:35:59.000000000","tz":120},"subject":"Blackify keystone.api remains","message":"Blackify keystone.api remains\n\nChange-Id: Ia5a57323d4f9be9d78d79c1bc33803cbe4fc1f41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a8be0dd30fdfc90885938df2d2c127d75d885df6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a8be0dd30fdfc90885938df2d2c127d75d885df6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
