)]}'
{"id":"openstack%2Fkeystone~924004","triplet_id":"openstack%2Fkeystone~master~I169bf258efa7e7f1cfa3f083a6e9ead1eba27a6a","project":"openstack/keystone","branch":"master","topic":"renovate","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-07-12 12:44:03.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"I169bf258efa7e7f1cfa3f083a6e9ead1eba27a6a","subject":"Blackify keystone.server","status":"ABANDONED","created":"2024-07-11 19:44:00.000000000","updated":"2024-07-26 07:54:42.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e3b3c97e148f2eca8d661f2253be8abb3c33f50b","_number":924004,"virtual_id_number":924004,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-12 20:16:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-11 20:48:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-11 21:37:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0df6170f2d49cf65ada8f872d58979489e6aada0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-11 19:44:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9fc518b6116c208be4fcaf0ed5f319c32cf8e624","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 20:48:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb90d5e1a64e491394eccbcb9bc782c9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2899408fc4bc4f1abfcd6797de7149ee : SUCCESS in 1h 01m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/431cae99ab2b4426a26249e825881f48 : FAILURE in 3m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dfd7a4cd3163336a8b95ddbce716c3fa56e5f704","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 21:37:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d6fea5a76d0e44ffaa224b93ba72fbaa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af72ababb781448fac4bcac1d1d95512 : SUCCESS in 21m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2148039e00f842d7baa719865a83b94d : FAILURE in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ebda507a8884944b280763dd9dacd9a : SUCCESS in 12m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d9e742d83c9415a9bfcba749bd4102a : SUCCESS in 10m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c1585c6c155b464a8bfae0625c59fd30 : FAILURE in 12m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/378c2704b414433bbe4aaecc1af9e036 : SUCCESS in 10m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/7e007023695849989c715785b66f7b21 : SUCCESS in 1h 16m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8c31b455cf54597ac8a1a584ebba6b7 : SUCCESS in 1h 39m 33s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/585cd7a6a0e64d489a46f12594c73325 : SUCCESS in 28m 11s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/c4eeac697df348bebad0e3796fae23e4 : SUCCESS in 27m 55s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0fdeb0eeec2746cba4404133b495cecd : FAILURE in 14m 28s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/a8290937dd5d424eaba4fa23b5b5be4c : SUCCESS in 30m 43s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/8aad3ddf2e674123afa18988fddc32da : SUCCESS in 23m 49s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/df5880502f4148d8940aac2e6ada1923 : FAILURE in 22m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0536b0c3f3444e76a76d6a2ab2637e5e : SUCCESS in 40m 01s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/69e76e7f704a4a53b04a7ad55773cddd : FAILURE in 33m 48s (non-voting)\n\nWarning:\n  Comments left for invalid file keystone/api/auth.py\n  Comments left for invalid file keystone/api/users.py\n  Comments left for invalid file keystone/assignment/core.py\n  Comments left for invalid file keystone/federation/utils.py\n  Comments left for invalid file keystone/identity/backends/ldap/common.py\n  Comments left for invalid file keystone/identity/core.py\n  Comments left for invalid file keystone/resource/core.py\n  Comments left for invalid file keystone/tests/unit/assignment/test_backends.py\n  Comments left for invalid file keystone/tests/unit/test_v3_auth.py","accounts_in_message":[],"_revision_number":1},{"id":"157134175f093b6e61a91fe4afe767c4ed7dd8d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 06:48:48.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"69398ef4255d2b5193d3754b440dc4d176e75c80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 07:52:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84006a6a3f6048e6b25c5f9bcb95bbc2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c01fc22ea40f49a9a3109bdc50a7b416 : TIMED_OUT in 1h 01m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/48f335a1d8b14b388852a29434525ea5 : FAILURE in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f3752fc9e6b8ec3821647d59019e2b0e3aea9ebe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 08:01:41.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"f7d35360a34c258a20608d9b06992083bbb6dda8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 09:05:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9728aeb444ca4f50baa5d30a3697b162\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f35724265af4dd585d20e80d85acc0e : FAILURE in 4m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c2123846e66e4b8099c6e704035e7dc1 : TIMED_OUT in 1h 01m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"51638a99a0f888543af72b35563640137466f41b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 09:29:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09ccf3c4bf6b453d8064d6f920dd5238\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4937b11fceac4418ad783fcc7da7f7d6 : SUCCESS in 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/babeb140549440c2af33c60dcba02013 : FAILURE in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b803cc961b6349c3bf29c0277bb65ce7 : SUCCESS in 12m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c52d894cf904b7784e4b668f1d60ee0 : SUCCESS in 10m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55070ae5c70a45c6875bc64b7e6940b9 : FAILURE in 16m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3090e48e50694312b3ba00354a0b0e37 : SUCCESS in 12m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/13372fa8f1dd4e649fb256d5041b5f4f : SUCCESS in 55m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/26ff434ebaab4a7b9df413763bc40e0a : SUCCESS in 1h 21m 06s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/109d320377854e68bc49352e1eabdc8c : SUCCESS in 32m 55s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/a1fa1a5a3745488b833329871f0914c8 : SUCCESS in 33m 15s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32fa092f72e9441bbc409fac14385ca8 : FAILURE in 16m 07s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/62e4d06779254f869d3f9f11aad22829 : SUCCESS in 26m 37s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/f7264d171364458dab73c672e265e100 : SUCCESS in 20m 27s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/4518dd201f5d40d39910e1dcd9660c58 : FAILURE in 23m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/904ad96301744461b15358e0188331ea : SUCCESS in 52m 09s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/5f387ae62d6449129e9aba51d2d1bdf8 : SUCCESS in 40m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b848aa32f7049870ce85c329082a2beb412fce6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 10:42:13.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ecfcb08268d1b7769e1973cb3c638d9843ed6cda","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 10:58:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/345328b63f0445c9b2f5c3f6bae95423\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a8bc687b3de4f39a1071ca22611c81a : SUCCESS in 13m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/02ab6b08f2cf4be5bd43e75d760722f9 : SUCCESS in 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"04d0a9af9b114b6897311373745373a327db873b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 12:14:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c90561c050f44d868ebbb96b5aee34e7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/397d460d5f314402ac5d0b6fe9d0ca44 : SUCCESS in 15m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adaa7f81f6734b858c58206731afd610 : SUCCESS in 7m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de6f7503aa144d799493a51a8e91abcd : SUCCESS in 13m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f034f42e6994473da4e02445a0bf03f9 : SUCCESS in 11m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/57310e0e0b4e4ea0be61bf7ed6cc5546 : FAILURE in 16m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c44279c3374d4fe1bdbd02f87b55b957 : SUCCESS in 13m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/375b6cd8cbb7422daf76ac39e6210a54 : SUCCESS in 59m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/861ecb0e64e343cc9b5a4e6005c47b52 : FAILURE in 1h 21m 39s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/60a7ac9a2c05451883157c48273406f5 : SUCCESS in 34m 36s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/c3db72afe4d24502a93a7e13bba09ef5 : SUCCESS in 29m 23s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b27fdfeeb2c64ed4b580a6109e24306f : FAILURE in 22m 39s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/9be2057872bc42ddbd0a5af44cd81b54 : SUCCESS in 31m 44s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/3ce233103bf84808be73c7e410130407 : SUCCESS in 21m 07s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/67a0133ad5014293bbf14d54caa64d41 : FAILURE in 14m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f452f1a26f0d4506a4d907ceaedf33e6 : SUCCESS in 1h 02m 26s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/8161759dbb924b25aae30230a0b516a0 : SUCCESS in 35m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d720c7f24906cad0ac8f6d7150df8de3eea326ec","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 12:44:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"46e8c5cf69a1d96de081707fa651eae1f97e7a13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 13:46:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4da029b570474c09830ac53501b8538d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/981dc344843f4090afe6868ccc49c87d : TIMED_OUT in 1h 01m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1a4bb1e230b349ffb49fdf91762f1237 : FAILURE in 3m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f0e4e06bf296269331c8bd29d82a6dfbb9d472ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 14:10:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00aae58cf56340e4a1493c1acffcb13a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/471a435f0b03437697076d49d066c8a2 : SUCCESS in 16m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47db597b399343339204036520549ecc : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3434fdc84e3d4dc196f3294a63ac8c8a : SUCCESS in 11m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/163a5b784aea40368bc1529c331b1ead : SUCCESS in 10m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7a940e34032c40d7a2daf5bbced90788 : FAILURE in 11m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f786076ab6d436789ae5737755524a6 : SUCCESS in 14m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/07c52d7b0c1d4ecd8641d725381d7b0b : SUCCESS in 1h 07m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d41b7fe1bad49b1a918bd008cd4421e : SUCCESS in 1h 23m 56s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/cba8fdcd85434e48aaa97f955d250321 : SUCCESS in 26m 33s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/2906b1436598418390a513999bd8b003 : SUCCESS in 30m 36s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da9434236e264213b3bb475051a77ae2 : FAILURE in 27m 59s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/d5f7cf0591da473bb965ff9eb1671433 : SUCCESS in 28m 49s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d988ef6ee55748dbad375d67b5b64922 : SUCCESS in 21m 49s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/0b679266b57b4fd8940651283c25cc07 : FAILURE in 13m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b71b80e8e5a4cd9970e306b9304647e : SUCCESS in 1h 01m 53s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/540d61cc12cf4ce59607e7b0d2f03178 : FAILURE in 45m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"753279ed403430785393ae7509a4af052ef858c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 18:36:13.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"29194d33f2a810131013e10f42b4592359d07361","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 20:15:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d78e787d2344067a37331aef2b7417b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad873851371b47f5bbd85a6ba55b37fc : TIMED_OUT in 1h 01m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d92c4f46436f4498951b272bfbac2701 : FAILURE in 3m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7999548866b5ee4c163f81827db775aa71d337ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 20:16:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c589e78dd004a8bb8731c2fd22d0148\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae52bfa02ae143748ed394eaa3f02a38 : SUCCESS in 14m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0e1e8c82025497b97f316316ca104ae : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9818b7e1cfee48ccbbaa1858f5612aa1 : SUCCESS in 11m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/15b578472fe543a6869c3ad6a36b34ef : SUCCESS in 10m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/81bceb6c7d344aefbb3268cbc0b3e36b : FAILURE in 15m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0c99b18df8745ee8f5f7234a6351b3f : SUCCESS in 11m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/88a25580f3a442a4a588a63c25bfbae0 : SUCCESS in 55m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d149a6d173a3441ea07d0bc291d4b392 : SUCCESS in 1h 32m 20s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/12269878b5e0445590355bbf28577725 : SUCCESS in 34m 01s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/5092810bb20e410bacf8bba4ae701222 : SUCCESS in 27m 05s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61f6d6d5b1d542f39b6716007502c2b0 : FAILURE in 26m 29s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/2fb806200ba242508932f859af77c8b9 : SUCCESS in 23m 20s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/aef0328ae5d34d8f876e6c2a2bb851f6 : SUCCESS in 19m 10s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/5aeaceeff66243abb571b5a50e9a96d6 : FAILURE in 12m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6ec47511a634bd9aefd96fd5974757e : SUCCESS in 49m 05s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/7c829fe86f124e47823f0410860ccbce : SUCCESS in 41m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e3b3c97e148f2eca8d661f2253be8abb3c33f50b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-26 07:54:42.000000000","message":"Abandoned\n\nLanded in other change","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4d14920dbd640d41df720f113836458d55c1df8a","revisions":{"d2573968d579c42e1e6c2ac3d31f4cf134230999":{"kind":"REWORK","_number":1,"created":"2024-07-11 19:44:00.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/04/924004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/04/924004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/04/924004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/04/924004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/04/924004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/04/924004/1"}}},"commit":{"parents":[{"commit":"449570b7ccd783c271579fe9a3c8a73a4b5b67bc","subject":"Blackify keystone.api remains","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/449570b7ccd783c271579fe9a3c8a73a4b5b67bc"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:23:44.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:23:44.000000000","tz":120},"subject":"Blackify keystone.server","message":"Blackify keystone.server\n\nChange-Id: I169bf258efa7e7f1cfa3f083a6e9ead1eba27a6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d2573968d579c42e1e6c2ac3d31f4cf134230999"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d2573968d579c42e1e6c2ac3d31f4cf134230999"}]},"branch":"refs/heads/master"},"65c759ea149369f2f7b1028be37236f56a908f6d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-12 06:48:48.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/04/924004/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/04/924004/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/04/924004/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/04/924004/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/04/924004/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/04/924004/2"}}},"commit":{"parents":[{"commit":"df43ff1f27816518987e9711ec9052eec7f51999","subject":"Blackify keystone.api remains","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/df43ff1f27816518987e9711ec9052eec7f51999"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:23:44.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 06:44:59.000000000","tz":120},"subject":"Blackify keystone.server","message":"Blackify keystone.server\n\nChange-Id: I169bf258efa7e7f1cfa3f083a6e9ead1eba27a6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/65c759ea149369f2f7b1028be37236f56a908f6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/65c759ea149369f2f7b1028be37236f56a908f6d"}]},"branch":"refs/heads/master"},"8107ca2f5420deb3fe84084e962aed751da97014":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-07-12 08:01:41.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/04/924004/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/04/924004/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/04/924004/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/04/924004/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/04/924004/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/04/924004/3"}}},"commit":{"parents":[{"commit":"a1b1bdda18c1ff6432326acb0fcc5294e0574259","subject":"Blackify keystone.api remains","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a1b1bdda18c1ff6432326acb0fcc5294e0574259"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:23:44.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 07:21:24.000000000","tz":120},"subject":"Blackify keystone.server","message":"Blackify keystone.server\n\nChange-Id: I169bf258efa7e7f1cfa3f083a6e9ead1eba27a6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8107ca2f5420deb3fe84084e962aed751da97014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8107ca2f5420deb3fe84084e962aed751da97014"}]},"branch":"refs/heads/master"},"1f4f7c3796ddf47637151c5928d07d0e55aaec21":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-07-12 10:42:13.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/04/924004/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/04/924004/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/04/924004/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/04/924004/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/04/924004/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/04/924004/4"}}},"commit":{"parents":[{"commit":"86af955db2e0f7f1d2593d0fd7b092017fb53066","subject":"Blackify keystone.api remains","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/86af955db2e0f7f1d2593d0fd7b092017fb53066"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:23:44.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 10:42:13.000000000","tz":0},"subject":"Blackify keystone.server","message":"Blackify keystone.server\n\nChange-Id: I169bf258efa7e7f1cfa3f083a6e9ead1eba27a6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1f4f7c3796ddf47637151c5928d07d0e55aaec21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1f4f7c3796ddf47637151c5928d07d0e55aaec21"}]},"branch":"refs/heads/master","description":"Rebase"},"4d14920dbd640d41df720f113836458d55c1df8a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-07-12 18:36:13.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/04/924004/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/04/924004/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/04/924004/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/04/924004/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/04/924004/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/04/924004/5"}}},"commit":{"parents":[{"commit":"a8be0dd30fdfc90885938df2d2c127d75d885df6","subject":"Blackify keystone.api remains","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a8be0dd30fdfc90885938df2d2c127d75d885df6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:23:44.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 18:35:59.000000000","tz":120},"subject":"Blackify keystone.server","message":"Blackify keystone.server\n\nChange-Id: I169bf258efa7e7f1cfa3f083a6e9ead1eba27a6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4d14920dbd640d41df720f113836458d55c1df8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4d14920dbd640d41df720f113836458d55c1df8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
