)]}'
{"id":"openstack%2Fkeystone~924007","triplet_id":"openstack%2Fkeystone~master~I2112eac5e412c1bba496744c6edc7811231f0eb5","project":"openstack/keystone","branch":"master","topic":"renovate","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-07-26 07:57:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2112eac5e412c1bba496744c6edc7811231f0eb5","subject":"Blackify keystone.protection tests","status":"ABANDONED","created":"2024-07-11 19:44:00.000000000","updated":"2024-07-26 07:57:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"528b0c592a1424d5ee4ef8ceb47110d56d7c775b","_number":924007,"virtual_id_number":924007,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-12 20:42:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-11 19:54:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-11 21:16:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9556803bb42cafbf8037b664d394e681483ddfec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-11 19:44:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d91ea950900ab0ca1f46646cd85dfd912e2eabf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 19:54:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afd117e359124331babc490d52665e19\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/836811aee29348cfb50b3c319887950a : FAILURE in 4m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/76f44b2f9b654eaa890af2648e6269d2 : FAILURE in 3m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"97fdd786019b01789aa71e709ba458b612e303cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 21:16:02.000000000","message":"Patch Set 1: Verified-1\n\n(23 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e4dc417bb784607a18f69c8e1f716e2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed49b940dc824e18bf6c690a4421fe31 : SUCCESS in 18m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f441e832a6a4e26891140da361f89fb : FAILURE in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1a3b2f451864c11aa0fec71a326ded1 : SUCCESS in 11m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/08841ee727e84f2ebf88e17b989108a5 : SUCCESS in 11m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2e4e944e0acd4866a1d3b4979950ca32 : FAILURE in 18m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c2af7fff951429294ee665faa2b205e : SUCCESS in 12m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/d6dc7d1582644d6f8f922b7723d185eb : SUCCESS in 55m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/879fbade94e7422b875c32c1b4cdf00d : FAILURE in 1h 13m 12s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/768ae4d90be54a68b090c4ecc892c537 : SUCCESS in 27m 54s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/59c900af7679460c9337f53d99b56e65 : SUCCESS in 39m 14s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12ba27d6ca1a40aebe64ef283103846e : FAILURE in 13m 59s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/44fbca857a30452293bb6f2d1a303598 : SUCCESS in 30m 26s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/31626f35303a43a6bf450299a387341a : SUCCESS in 17m 49s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/8db26cdc157c4902933c7f1d4ec493ff : FAILURE in 21m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbd302e0409840cdb453cc67ec6317bf : SUCCESS in 52m 56s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/cb348fa676ec4e5db21bd2afef836f05 : SUCCESS in 28m 29s (non-voting)\n\nWarning:\n  Comments left for invalid file keystone/api/auth.py\n  Comments left for invalid file keystone/api/users.py\n  Comments left for invalid file keystone/assignment/core.py\n  Comments left for invalid file keystone/cmd/cli.py\n  Comments left for invalid file keystone/cmd/doctor/__init__.py\n  Comments left for invalid file keystone/common/password_hashing.py\n  Comments left for invalid file keystone/federation/utils.py\n  Comments left for invalid file keystone/identity/backends/ldap/common.py\n  Comments left for invalid file keystone/identity/core.py\n  Comments left for invalid file keystone/resource/core.py\n  Comments left for invalid file keystone/tests/unit/assignment/test_backends.py\n  Comments left for invalid file keystone/tests/unit/test_v3_auth.py","accounts_in_message":[],"_revision_number":1},{"id":"48f2ba8a06c3f5c471e3538917dd3e47ddc5ac38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 06:48:48.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3de44df1f7e8b504d61243ee6bbdf87128e7de9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 07:02:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8369855f7bf141859de7ddf82d734f52\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e2ba845b975642b3a785d1869e753029 : FAILURE in 4m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5c2640398da648d1a17dbecc8922a125 : FAILURE in 4m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"56b9763a039282bf9e4e4c15561c863e677c3c6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 08:01:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5da786fb272bf5b18835350a13baaaccdff96e81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 09:06:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa0b4a40ea9f444c8e345a7b689010c0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/25e592919c364252bc26dea906d46371 : TIMED_OUT in 1h 01m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/01fb9edf31b54afd8b1f4103316fbff1 : FAILURE in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"db76e743a7ce7c60a534dcf2f0a4d530c11906c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 09:43:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce8427428589452b8a5aeb7f218ee2a6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/83006d7fa3634aaca1520b96eb85e0ea : SUCCESS in 4m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fc9402a1c3146a098724e24a57c3588 : SUCCESS in 13m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa2a656f5faf4a66b0b442b6151368ef : FAILURE in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9dd39b7130e34ca7a46c8b40eb4139e1 : SUCCESS in 11m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f2edd87d454496c9ee1d0ab202ff64a : SUCCESS in 11m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f98c427df4f94640bbe70f260514cb37 : FAILURE in 14m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41229e4dc55f4e4e80f093bb7ddf4897 : SUCCESS in 11m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/3a3ca79e437a4354bfb4c458b9bafb17 : SUCCESS in 56m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb8daf68448446cf8106fadd845a4ee6 : SUCCESS in 1h 32m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5086aa5e40c437c9b6316de0b9537ae : SUCCESS in 11m 12s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d341834c8ed54f45884c1515e205a487 : SUCCESS in 28m 49s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/cc0c2e4d979b4900bdeb6d96730852d3 : SUCCESS in 30m 54s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/442c2471aa73489794805d1826dd97d6 : FAILURE in 25m 16s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/8aa70fe9efd34046a81527353e7d23ee : SUCCESS in 33m 20s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/3a7088392a91441592669c90892fc093 : SUCCESS in 11m 42s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/75107979b34348c5b0e006b5b3a7655f : FAILURE in 20m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/debad0d86d95437cb48747025ea685d7 : SUCCESS in 1h 03m 40s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/6dfdd394622b4385bfd27c1c7a295310 : SUCCESS in 35m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6bc9ffdcf3b9f48409bcfc6077c299e358d6c417","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 10:42:28.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7f2b2d95574c66df539e5c55e81d63d5351345fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 10:58:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/484fa1efc2dd4ca3a3f25bc3b0222515\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9cd25eed697f4305bda7de79eccea968 : RETRY_LIMIT in 2m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/59ecd60f7a4f45eeaa8928a06b647e8d : RETRY_LIMIT in 2m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"33a7c28bb8d2fd752fb9f72344352493082bd5e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 12:35:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3cada66bf03e4bc0baa47efed294bc8d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/343413916a9c4cd4bdf505e16d7cf5f1 : SUCCESS in 6m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a08315fb00fd4ecbbe07ff3285c6aff7 : SUCCESS in 14m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01929ad0e5c9447eb948212122f88df4 : FAILURE in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/102b3b27a0b54d329affa13b5835e59d : SUCCESS in 13m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/935d5294e03b4335b16df96e5ce30f60 : SUCCESS in 10m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a15e8623bbf44789ba347fd1c40495ca : FAILURE in 15m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ed7432bdaa942eda4bea72da52923b4 : SUCCESS in 11m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/e6dd398f11a14e7b9b4b5aa04530bd64 : SUCCESS in 56m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf0f8e4bc1964042987ab8eea9342006 : SUCCESS in 1h 39m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/455aa47fc941479798243b471581dfc5 : SUCCESS in 11m 37s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/fc3b7b3cf41448c5b5ffd7d0a49c7485 : SUCCESS in 29m 29s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/6e0293b6f4874375af9c24b915505ca2 : SUCCESS in 43m 24s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7165b01596940ea9bf4d6f13b510c1e : FAILURE in 22m 52s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/a25c6c88ba3c4fb2afc746092569a921 : SUCCESS in 21m 19s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/5273732554844be6b18c3656a13d8340 : SUCCESS in 11m 43s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/e1c35133b1ce49e09e4c2c48fa5f40a9 : FAILURE in 10m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/082890c2b6a446339691fec7a013fb15 : SUCCESS in 55m 17s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/357b5d31961b472daa6a586507a38f1e : SUCCESS in 35m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1201818209d1689e34bdb7b69b37068fd4f92e51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 13:20:36.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"f13fefc8f8501d1e5cb1d051f76545c482ac5295","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 14:24:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/023b4d1c1d214934a4ea5a4d63093bee\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e98af2ca4adf44ca811420dafe32a3fb : TIMED_OUT in 1h 01m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/10da11ee7baa4e81ab27b3907188dacf : FAILURE in 3m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b2b5eb526fc3efba0b22dea5a50ce9fd3d407422","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 15:15:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9b84353bcfb4e8b88154e5b065ff195\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d90191dc472b4e08aac0aaa484993166 : SUCCESS in 4m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb13d158374f42219b0b5742c1626a64 : SUCCESS in 14m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b5ddb55211d4f29bd3945cf8478024f : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54f35a2d09894413908a593bc32b754d : SUCCESS in 18m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95e45b03975c458db6116bef5d7ba50f : SUCCESS in 11m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a6d88f46d81344c186bddb5c427d93b7 : FAILURE in 22m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1501781ba6564cf99c4b4713ff74eaa3 : SUCCESS in 15m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/9a531a5c958e49c39b973ba3fadc4f5b : SUCCESS in 1h 01m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8196dbb377a14f0b90eff5dc3b0e1890 : SUCCESS in 1h 50m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc306d47ba3e4ac0b6b98de004864690 : SUCCESS in 9m 22s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f8d3d7e27e664cbf80fcd4fe88e26900 : SUCCESS in 27m 42s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/3245b8109df04ea0833390d6d642cd91 : SUCCESS in 43m 02s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/797b0702ce854a379791e80f18d81381 : FAILURE in 25m 48s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/738f8f923d3345afb93233351bd2ff4c : SUCCESS in 25m 29s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/81963b1169a046ca8ec4fa41fd2206d4 : SUCCESS in 11m 17s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/33814a281edb47049f4f3b92315752b7 : FAILURE in 23m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/106720dd7c3647e1a4d7a9e30f0e5128 : SUCCESS in 1h 05m 26s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/2545aa6266e744bf9b2c49677c5ef645 : SUCCESS in 44m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c2ec34a2ccc75a5383ea3cf887e82653e79d493d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 18:36:13.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"eee36322d1bfd32da5709b7a1bd5d49e2502016e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 18:57:05.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb2bc2564c6f46159de4b6feb0af4f35\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c3d994365d514c8d95113a9e79862d79 : FAILURE in 4m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/95aa22a1420a4a79bc296a435c6b4931 : FAILURE in 5m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5c0f2539ca521c83b4432b00a1f1a7a71f41572d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 20:42:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7acf3c104c6442a885b60a4f2119831e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ad150955fd6f4bde8d2bf3c739fdaf86 : SUCCESS in 5m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c216df57b9b4babb7f901947dd48430 : SUCCESS in 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2af040a59a8c4ceeb0602465ad702a67 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/293681b801484919af62e07ed48ef6fe : SUCCESS in 12m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c09ab5c42b04fb8a90d4f1814799521 : SUCCESS in 11m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7afe9d6e7d1f45d780c0fd5f8166162e : FAILURE in 12m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e09123c64fe645819df99e6fadd7a512 : SUCCESS in 11m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/913942b8abf948649348a3aa43b58d9f : SUCCESS in 1h 14m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab6d604f5dcf4cfebb05519384e506e7 : SUCCESS in 1h 33m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/340a0589efab4dad80648f15ab47f3ae : SUCCESS in 13m 14s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/c3c36e1eec8c4a27aea53db7813ae655 : SUCCESS in 28m 22s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/1c42f6d388654a02959e2160bf52b502 : SUCCESS in 38m 46s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a79299e45ebd40028a75408701ec1ed3 : FAILURE in 22m 53s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/676f9c59fbe844f9a5620a57e49b3c35 : SUCCESS in 31m 41s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/8b3d76a9a5884383895e1365c57c19e0 : SUCCESS in 11m 02s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/393ac866d02241ad92fd70b70ae0e738 : FAILURE in 12m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d854f998d5714402aedccdb79e0833cc : SUCCESS in 55m 09s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/2faff36e586b422faf14eded1d887117 : SUCCESS in 35m 58s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"528b0c592a1424d5ee4ef8ceb47110d56d7c775b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-26 07:57:41.000000000","message":"Abandoned\n\nLanded in other change","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"5aae17c6702b649c804bd91cb9f4ae326e6db7b0","revisions":{"e157f24cd81dde3fd56da410ffb8fafa12701b8f":{"kind":"REWORK","_number":1,"created":"2024-07-11 19:44:00.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/07/924007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/07/924007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/924007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/924007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/924007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/07/924007/1"}}},"commit":{"parents":[{"commit":"7ae4587f08b527baf7a717c4ec7f880cdb3b0f5b","subject":"Blackify keystone.common","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7ae4587f08b527baf7a717c4ec7f880cdb3b0f5b"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:25:27.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:25:27.000000000","tz":120},"subject":"Blackify keystone.protection tests","message":"Blackify keystone.protection tests\n\nChange-Id: I2112eac5e412c1bba496744c6edc7811231f0eb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e157f24cd81dde3fd56da410ffb8fafa12701b8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e157f24cd81dde3fd56da410ffb8fafa12701b8f"}]},"branch":"refs/heads/master"},"d2a6de8c7ceb873b7cf89076faf747b357e2e910":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-12 06:48:48.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/07/924007/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/07/924007/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/924007/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/924007/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/924007/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/07/924007/2"}}},"commit":{"parents":[{"commit":"47d4a13443b7bc9d135bdc9158f88665244c42af","subject":"Blackify keystone.common","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/47d4a13443b7bc9d135bdc9158f88665244c42af"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:25:27.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 06:45:29.000000000","tz":120},"subject":"Blackify keystone.protection tests","message":"Blackify keystone.protection tests\n\nChange-Id: I2112eac5e412c1bba496744c6edc7811231f0eb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d2a6de8c7ceb873b7cf89076faf747b357e2e910"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d2a6de8c7ceb873b7cf89076faf747b357e2e910"}]},"branch":"refs/heads/master"},"ffb42cfe054e2d6cd1c83baaccdb9b9b90b84412":{"kind":"REWORK","_number":3,"created":"2024-07-12 08:01:41.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/07/924007/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/07/924007/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/924007/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/924007/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/924007/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/07/924007/3"}}},"commit":{"parents":[{"commit":"6d7b211f86d17bdd7f5b151089e218231cc14dd0","subject":"Blackify keystone.common","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6d7b211f86d17bdd7f5b151089e218231cc14dd0"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:25:27.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 07:51:14.000000000","tz":120},"subject":"Blackify keystone.protection tests","message":"Blackify keystone.protection tests\n\nChange-Id: I2112eac5e412c1bba496744c6edc7811231f0eb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ffb42cfe054e2d6cd1c83baaccdb9b9b90b84412"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ffb42cfe054e2d6cd1c83baaccdb9b9b90b84412"}]},"branch":"refs/heads/master"},"aeb3becc48983976665239ab9f91d64c049cb4e8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-07-12 10:42:28.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/07/924007/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/07/924007/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/924007/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/924007/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/924007/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/07/924007/4"}}},"commit":{"parents":[{"commit":"d1f0000dc572165b069fe0339507e64f0a129b6c","subject":"Blackify keystone.common","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d1f0000dc572165b069fe0339507e64f0a129b6c"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:25:27.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 10:42:28.000000000","tz":0},"subject":"Blackify keystone.protection tests","message":"Blackify keystone.protection tests\n\nChange-Id: I2112eac5e412c1bba496744c6edc7811231f0eb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/aeb3becc48983976665239ab9f91d64c049cb4e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/aeb3becc48983976665239ab9f91d64c049cb4e8"}]},"branch":"refs/heads/master","description":"Rebase"},"51a2d389becd993b785512a244e92eada9db45cd":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-07-12 13:20:36.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/07/924007/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/07/924007/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/924007/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/924007/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/924007/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/07/924007/5"}}},"commit":{"parents":[{"commit":"d97d4830de823bc643aaf0f1979ac5650abe9505","subject":"Blackify keystone.common","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d97d4830de823bc643aaf0f1979ac5650abe9505"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:25:27.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 13:05:45.000000000","tz":120},"subject":"Blackify keystone.protection tests","message":"Blackify keystone.protection tests\n\nChange-Id: I2112eac5e412c1bba496744c6edc7811231f0eb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/51a2d389becd993b785512a244e92eada9db45cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/51a2d389becd993b785512a244e92eada9db45cd"}]},"branch":"refs/heads/master"},"5aae17c6702b649c804bd91cb9f4ae326e6db7b0":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-07-12 18:36:13.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/07/924007/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/07/924007/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/924007/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/924007/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/07/924007/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/07/924007/6"}}},"commit":{"parents":[{"commit":"921fae5ee8dbdc8fbdbe51ed5280763314a0f1d5","subject":"Blackify keystone.common","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/921fae5ee8dbdc8fbdbe51ed5280763314a0f1d5"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-11 16:25:27.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 18:36:00.000000000","tz":120},"subject":"Blackify keystone.protection tests","message":"Blackify keystone.protection tests\n\nChange-Id: I2112eac5e412c1bba496744c6edc7811231f0eb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5aae17c6702b649c804bd91cb9f4ae326e6db7b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5aae17c6702b649c804bd91cb9f4ae326e6db7b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
