)]}'
{"id":"openstack%2Fkeystone~924085","triplet_id":"openstack%2Fkeystone~master~I7246d4e68386e9f72559e8948c475e5916359922","project":"openstack/keystone","branch":"master","topic":"renovate","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-08-21 16:56:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7246d4e68386e9f72559e8948c475e5916359922","subject":"Enable mypy","status":"MERGED","created":"2024-07-12 17:02:44.000000000","updated":"2024-08-21 16:59:21.000000000","submitted":"2024-08-21 16:56:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924085-renovate","meta_rev_id":"20b1c1a5c3b287a9aed4e81ce4a6decf792c3162","_number":924085,"virtual_id_number":924085,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-21 16:56:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2024-08-07 14:54:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2024-08-21 15:01:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2024-08-21 15:01:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-12 17:07:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-12 18:15:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-07 14:54:42.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2024-08-21 15:01:48.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"86e16b057565101a5171f24f0f46886aa3b2067a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 17:02:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a9af69e4aec6c5d6551fcb3d9dbfb3a94c9f7f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 17:07:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a4ea27b95a04bcda39d3935907deae8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/45de44a722794ca0a3d352bc7825c0c6 : FAILURE in 3m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/76d6ae90c10d435bb653df76ffb3280c : FAILURE in 3m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e63aab76823d7de29a3de100438502a488b93a96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 18:15:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b607651ca77944a6a127e03c90c7500c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/091eedaed56045da99fe20d062e689c6 : SUCCESS in 4m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d7c6455767d4e51969a2314777f4d1f : FAILURE in 12m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a4bcb955bc7445092b377759c740a7b : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9b0265c2e7148508b2ad8f812bdb637 : FAILURE in 12m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee54bb77a9bf4cfc93fd393d972c3a81 : FAILURE in 10m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/07cd4ddd34f24b4493480624966a8623 : FAILURE in 18m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d14cf67f0cb84cd18193f9ffc7648c42 : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/37b9e320610845e1a650ec07675e1d18 : SUCCESS in 1h 01m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5afdd9f72f714985b891d30620dc54d2 : SUCCESS in 1h 11m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e5fd7bed1f143a2b495a3b196f211ef : SUCCESS in 10m 50s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/76dd68866d0c410d9df62a4a5a73cbd3 : SUCCESS in 19m 24s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/074b12caf0114e6e907354c0ba971959 : SUCCESS in 29m 04s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4e9b02e98d04650b6b3056f25637add : FAILURE in 18m 00s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/414a8daeb6e841779141071c9de2e2a7 : SUCCESS in 30m 20s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/867cb40972454952ba308be21e9fc893 : SUCCESS in 20m 33s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/aaec7558f6504783b25fccb063271561 : FAILURE in 22m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b550a2705d614493b3d4cd848b083cb1 : SUCCESS in 52m 10s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/0329c6ca8b12492a9824b4013d7199d0 : SUCCESS in 44m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad97a9d614e746e95f7353bc95cc515098366f8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-12 18:36:13.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d11cbdcf83e2ca196e6bdfc44f104617bd9f0b99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 18:56:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffe88553048742adbf2948262768ff8d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bedf51adb2ed4b9a939eaec73807db1a : FAILURE in 4m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/77751d3f9077400fadc8c56c68d98b09 : FAILURE in 3m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d5516b5d39696f8ec9027b681ba62d503a756ff4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 20:28:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3987b9d170e14b368c0a7aee6969a883\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f0f1352817c41df8d855dc5c77a0812 : FAILURE in 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdc42d66442f4bf8bf27bcd3be287be1 : FAILURE in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c6856c0dc7a48f191ddcc33564fbbc8 : FAILURE in 10m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6205baa967a64a8c860aaa83e4c448b6 : FAILURE in 11m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0482563511344689a229df5fb052b4f4 : FAILURE in 12m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af27d8e5be9d4434a9cde581202658de : SUCCESS in 12m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/0afea3d5ffe348468adf705fba5b55b7 : SUCCESS in 1h 00m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f9494241530c447a8c26a8e35100d3eb : SUCCESS in 1h 44m 15s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/9a6e3cd7c40c4df4ade8dfb0017aeda0 : SUCCESS in 32m 43s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/a22d06985a974570b0a7c876e737b772 : SUCCESS in 40m 52s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a7981e0670349a8813249aa573fac17 : FAILURE in 24m 17s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/959bad82c5824b5796022039097fdcfd : SUCCESS in 30m 36s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/612cd3f06fc646f69b0bbcaace02cc0c : SUCCESS in 18m 09s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/bbe35c57f8af47bbb7f13b277a4fbcac : FAILURE in 23m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/591dcf9529ad40de85e1ae6bda97b762 : SUCCESS in 1h 01m 33s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/61853f33c2734fbb9f4938e780625777 : SUCCESS in 40m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"62ed3f1714acdd1a06152cb6c57ac6a804cfd60d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-19 07:44:39.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d09ed03c40f99f5a3fb99700f70e931d8d17c05a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 08:00:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1c285fd566743d7b8d4f7fc1525414c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/693b6cb89dc34c2c828b500ed13ad895 : RETRY_LIMIT in 2m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fa2ec5ba250348339677046f23b06f84 : RETRY_LIMIT in 3m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"54d0232a5886f208887a51de25a2f97164acfe82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 09:27:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffab19c25ef84165930ae917a99e2307\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/954b2f1f4a5a4895b1dc6cab0afd100d : FAILURE in 12m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f550bc299ab4603b1848632891160c7 : FAILURE in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9917cfa813034e5795a5c85f02af23a8 : FAILURE in 13m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e79ac8d6087440489c477db769b294fe : FAILURE in 12m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b69c338f6a1742e0a5bcb93741b0812a : FAILURE in 14m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdf3d37ee7f3435a8670b1b81bed0080 : SUCCESS in 12m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/588157b155a24e73b8369b770fa7e5bb : SUCCESS in 51m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2774e992f6be4a22a7c4c26b760b3efa : SUCCESS in 1h 36m 04s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b3a07d0ee17d49ac8e481856ccebe6a8 : SUCCESS in 31m 12s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/6b3705ad9dfd418ba4ba173894cc8430 : SUCCESS in 41m 25s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c63bfc0aabe4fca892ac29b54f9a59a : FAILURE in 14m 24s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/b934f990d1ba47ba8fdca71aa196d2f6 : SUCCESS in 26m 43s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c63cf5056bf74f359daf35ba44ad1abd : SUCCESS in 19m 35s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ac52e7afe85e462b85d0c92ee60ddb4a : FAILURE in 22m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/163b73122cf245ba8fea850ee60e0653 : SUCCESS in 55m 50s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/75da4fb6af6f444f925ae038842e965d : SUCCESS in 30m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cd7aa2a3d727c24e3f7ef971e0274da11a1903ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-19 09:53:50.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"bd77e17a6329ed9c1750876521ecf827270476af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 10:12:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/768959d94e2e47d985773ee0a0819b50\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3d6f460084b44f95882209213dd3d785 : FAILURE in 10m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aef276f26e0a483ea570a49cbd8c56ba : FAILURE in 10m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ce3101b8d5758726b4eaa4288723687b81701282","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 11:43:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4f4d8075c34418eb2fad2ae330455ab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c6899e2193c40669835594c468394b6 : FAILURE in 15m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94195d34d0fd4d51800345056d7dff62 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5814ac88f1af4fd193518fec115a7e4c : FAILURE in 14m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bf1757852dda43c48e17f6a391504075 : FAILURE in 10m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e71ae050fd6c49eeac4b0fbe4c508220 : FAILURE in 15m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b221b2bf7544775b29b6cfaf22b8645 : SUCCESS in 12m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/7a7407a40760446cb318733ba302378c : SUCCESS in 1h 05m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/63072e78b28041288e9eaa8921c16292 : SUCCESS in 1h 42m 20s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/0add662c98fe40bcac2ebbeb263061b7 : SUCCESS in 30m 08s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/103b87d8c70f4caca2e5ce39dc52768b : SUCCESS in 28m 13s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e44ae5dfa104db6aa04454f5ae0dbc8 : FAILURE in 18m 40s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/639811a76bad40f3b9b6bd1fccf9c1eb : SUCCESS in 20m 58s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e526b9fa5e60445caea83f44d2812731 : SUCCESS in 22m 21s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/62a44ec1c76348deb2b3606a15694d7f : FAILURE in 22m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5caa1a02f98140e3b0f011f717425d4e : SUCCESS in 56m 17s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/c286b12fdfe74bfb81c819bd95ef83ac : SUCCESS in 44m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7f495aa174e32600c4f79bcb288845e03b522cf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-26 08:11:11.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"230017fde51df9c4a29ecb3fa4766ddccf1a49bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 08:22:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/752f23c5510745e18c0d231a604b1030\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a9731eca06042ed9c41b453da778de8 : FAILURE in 10m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cc6c344f5f7645519bba4d447f7650c1 : FAILURE in 9m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d8a68ce60659740ab1a6f55b9a85422b515ecaef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 09:45:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37b12e92a7a44ddb9ac26d08d3c3ff07\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a9d65803fac4b1aa8e140a1461e9bf5 : FAILURE in 15m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/128418f56e334d35ac205dccf1789a45 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ca64a1b9bc544ec9748e34032281224 : FAILURE in 11m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7244101f4ed54848a6cacb8e9ea8a2ff : FAILURE in 10m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e566698985344785a48b7bbf0f811241 : FAILURE in 13m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39473be642d04115a51080f2a759efb2 : SUCCESS in 11m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/96a2af4c9184420a915f58afba425d2d : SUCCESS in 55m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d8eba58a93b645218a535d440841b748 : SUCCESS in 1h 32m 34s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/7920523440eb489aaccb508c60a7f859 : SUCCESS in 34m 04s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/ed1b1d9f02ba455b99bc5a91f31ab0f1 : FAILURE in 18m 55s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfb8bbe81a2c4d6a96135200201befc6 : FAILURE in 22m 58s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/151f7d8510334570aa4be3758190c255 : SUCCESS in 21m 06s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/7254ef16699240c98259503927baadb9 : SUCCESS in 12m 34s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/a34f3b0ac13743dfb22574c7c3e8bfc7 : FAILURE in 12m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d4e75f3e70741cc9acf764c168b67a0 : SUCCESS in 1h 01m 49s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/87adf58addde4e31b8d006d10a69b9b9 : SUCCESS in 44m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f09abfd83f32943a0ae9973bd65167011793da1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-26 09:51:30.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"cf788ea71981a2722b7448f307c21e33190deb37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 10:34:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9226a51188db4464b1e5bb663ff00a69\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/55ef1d152b634a8ab9a92f750a7e81a4 : FAILURE in 17m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1c19c35886fc4ef3b2f7a71454a3ae15 : FAILURE in 40m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4a8c1c79a72130cb4fe2457b77fc7e734f5b360d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 11:31:06.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c793aacdcab34a48abd1a26936033cf2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82cf6a116ebb4f549666b4b72d94d1fe : FAILURE in 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73b4be242b1344b5b6938b4478e3ec8e : SUCCESS in 7m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c62f17b525754cc3b85fcbc50fb55bb8 : FAILURE in 15m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e8f47955a25148f183c971476752981c : FAILURE in 10m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fd500944ff4a40dfae67747997cbd091 : FAILURE in 12m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/578b95681a4b40b9a58386681fe8faa5 : SUCCESS in 12m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/c3404a4a474e4fac8df001417e790525 : SUCCESS in 1h 02m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7bb086d01d77440f998e0db43bcb9ca5 : SUCCESS in 1h 35m 31s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/60bc6d55fadf492187d82fc244c437c2 : SUCCESS in 22m 06s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/615c855ad52b49b4858b9908f1175f96 : FAILURE in 9m 49s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/324fbf8655b442e7b48f3d9eba31c01e : FAILURE in 25m 39s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/66e0d9073da9432fad012dbc44c3d0c4 : SUCCESS in 29m 49s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/321762bcee9d4f149ca1035468c2fa52 : SUCCESS in 20m 15s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/2d4d680823a54b209edb95518b10ebc3 : FAILURE in 17m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4a25271462f445d93aaa20989815268 : SUCCESS in 53m 15s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/4f9ae00911834f1fb3ec3e23c7685a97 : FAILURE in 34m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"499aaf630304c28e3280847fad92584687301253","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-26 12:09:12.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"abd211108b56dea7a7e536f583a3abc711a5ad8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 12:20:26.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dc9a393492d4fbcadc86028be38d635\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b57a5623ad7b48beac259c0de2c2407e : SUCCESS in 10m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2f392e51ec6a437793b6bb197fcd0d1e : SUCCESS in 9m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cd2bb416bb144c3f602737e907e046a0179558bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 13:44:30.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ff49e38f06c4305a74ec0dd70adf45b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac6a9ddad6724601b0dc8efb8f3e826a : SUCCESS in 13m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d97433775624eac9c26f54fdb189938 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8cf7271ff8e4e178f5910491f6936e3 : SUCCESS in 11m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c7af4563c5a4b4e8596416fb5617b13 : SUCCESS in 9m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1bb0913a43664794ab41a9675acadb7e : FAILURE in 13m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d19d401215348ce8f38bbe1bc97f089 : SUCCESS in 11m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/ebd995737b35418b928d665384c79f47 : SUCCESS in 1h 08m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d066b6e2ebbb43658d9d00b1e6d68078 : SUCCESS in 1h 29m 08s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d305992d17dc41a7b67d62c15de5d60d : SUCCESS in 27m 53s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/9bdf796bc7eb49bf93ae21f838552070 : FAILURE in 8m 21s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42d9403e8c3240b8886f5f3738493bad : FAILURE in 16m 07s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/6c5243fac7444ff281bfc269ccbe85e2 : SUCCESS in 25m 04s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/233631c18bcb4ebda7fe4644201f55d7 : SUCCESS in 8m 55s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/0c2eda673e3a4586b81d2540176b7180 : FAILURE in 19m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bb13f65576f45a3bff20de016e3c84a : SUCCESS in 52m 33s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/1acc0eb1f2714e68943f2e961d994db7 : SUCCESS in 35m 23s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d30aaa187e0418935f7f9419dca83c741c57385e","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-08-07 14:54:42.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"767f2ee7a47e7e9b6f3386d9d6a3b85c1c8b8e8c","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-08-21 15:01:48.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"9483560f0e8f378c60209407c2160cc8731a68e4","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-08-21 15:01:52.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"312fbc58e0b5cbff6def93000d55fd6a57588931","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 15:02:39.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"c0cffde6a66c091b85b88b6309f3819957133ff6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 16:56:31.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ac97ed3a577497b911bbe5da1d173d4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbb1898a0be44fcebc903dee29fc0f91 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd677968b83a4ff7879ca7f10811c786 : SUCCESS in 12m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3dca0f5f61204b078e8e09738a1ba49e : SUCCESS in 10m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a2157db65874053ad9622f5f3928914 : SUCCESS in 11m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/3cf0f9a0e62b427180d1d0382e5ac1c7 : SUCCESS in 53m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a449430ef0544569a53eab0e1a4e50a : SUCCESS in 1h 30m 07s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/6118ac7d32814174a66392e3e3d27128 : SUCCESS in 24m 39s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/23e176cc4760474f8960d2402036953f : SUCCESS in 26m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76d9ea3e91b040adb6789c2f0282bf02 : SUCCESS in 52m 31s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/0e457406060b4f20b070c5bd411a4149 : SUCCESS in 39m 01s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"488058dcd51a10f95d46e8cae68505b23320ad40","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 16:56:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"20b1c1a5c3b287a9aed4e81ce4a6decf792c3162","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 16:59:21.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a46fee1d369941e283dc0b4b758c2c48\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecb3904901864e36ac9d70cedc55bf6a : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"80db930895be9151f522642756cd458f280f263e","revisions":{"cbca594869b389731fd0ba7df9ce1662215b7603":{"kind":"REWORK","_number":1,"created":"2024-07-12 17:02:44.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/85/924085/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/85/924085/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/85/924085/1"}}},"commit":{"parents":[{"commit":"14c5d086acf25372a8d7c687ac9a8e9f0095076c","subject":"Move bandit to pre-commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/14c5d086acf25372a8d7c687ac9a8e9f0095076c"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 17:01:42.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 17:01:42.000000000","tz":120},"subject":"Enable mypy","message":"Enable mypy\n\nEnable mypy and fix/ignore found issues.\n\nChange-Id: I7246d4e68386e9f72559e8948c475e5916359922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/cbca594869b389731fd0ba7df9ce1662215b7603"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/cbca594869b389731fd0ba7df9ce1662215b7603"}]},"branch":"refs/heads/master"},"ac0c4d2a84bec5f8900d28f1ef32b4ace2b33ad5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-12 18:36:13.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/85/924085/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/85/924085/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/85/924085/2"}}},"commit":{"parents":[{"commit":"fd30a6747c3fc10ae1fe5f2f801ec9befe98513e","subject":"Move bandit to pre-commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fd30a6747c3fc10ae1fe5f2f801ec9befe98513e"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 17:01:42.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 18:36:00.000000000","tz":120},"subject":"Enable mypy","message":"Enable mypy\n\nEnable mypy and fix/ignore found issues.\n\nChange-Id: I7246d4e68386e9f72559e8948c475e5916359922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ac0c4d2a84bec5f8900d28f1ef32b4ace2b33ad5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ac0c4d2a84bec5f8900d28f1ef32b4ace2b33ad5"}]},"branch":"refs/heads/master"},"bfd26f8453fd7fae56d71f5d3438b3ed4fdd399f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-07-19 07:44:39.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/85/924085/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/85/924085/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/85/924085/3"}}},"commit":{"parents":[{"commit":"d97c851d4f5b97fef7e689e2c7e27eaedbd94e52","subject":"Move bandit to pre-commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d97c851d4f5b97fef7e689e2c7e27eaedbd94e52"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 17:01:42.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-19 07:44:21.000000000","tz":120},"subject":"Enable mypy","message":"Enable mypy\n\nEnable mypy and fix/ignore found issues.\n\nChange-Id: I7246d4e68386e9f72559e8948c475e5916359922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bfd26f8453fd7fae56d71f5d3438b3ed4fdd399f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bfd26f8453fd7fae56d71f5d3438b3ed4fdd399f"}]},"branch":"refs/heads/master"},"3941d0e879877276ca9955a009cba12d8aa13331":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-07-19 09:53:50.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/85/924085/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/85/924085/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/85/924085/4"}}},"commit":{"parents":[{"commit":"2e42e03da772ea40a4ef26c467471fe293de947a","subject":"Move bandit to pre-commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2e42e03da772ea40a4ef26c467471fe293de947a"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 17:01:42.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-19 09:52:27.000000000","tz":120},"subject":"Enable mypy","message":"Enable mypy\n\nEnable mypy and fix/ignore found issues.\n\nChange-Id: I7246d4e68386e9f72559e8948c475e5916359922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3941d0e879877276ca9955a009cba12d8aa13331"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3941d0e879877276ca9955a009cba12d8aa13331"}]},"branch":"refs/heads/master"},"4f17699bd0c33505aa8aef194382ad906dea957d":{"kind":"REWORK","_number":5,"created":"2024-07-26 08:11:11.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/85/924085/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/85/924085/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/85/924085/5"}}},"commit":{"parents":[{"commit":"4035ce1d22c00cf6d10d8f54dcaad1ef171b2e68","subject":"Re-join the strings after re-formatting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4035ce1d22c00cf6d10d8f54dcaad1ef171b2e68"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 17:01:42.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-26 08:10:55.000000000","tz":120},"subject":"Enable mypy","message":"Enable mypy\n\nEnable mypy and fix/ignore found issues.\n\nChange-Id: I7246d4e68386e9f72559e8948c475e5916359922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4f17699bd0c33505aa8aef194382ad906dea957d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4f17699bd0c33505aa8aef194382ad906dea957d"}]},"branch":"refs/heads/master"},"1cec0e13e3e06081f3fdedf36d04597ef69ec962":{"kind":"REWORK","_number":6,"created":"2024-07-26 09:51:30.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/85/924085/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/85/924085/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/85/924085/6"}}},"commit":{"parents":[{"commit":"8416b72bff9d12f7baf097eb8efc9f7f5ddf8212","subject":"Re-join the strings after re-formatting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8416b72bff9d12f7baf097eb8efc9f7f5ddf8212"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 17:01:42.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-26 09:51:19.000000000","tz":120},"subject":"Enable mypy","message":"Enable mypy\n\nEnable mypy and fix/ignore found issues.\n\nChange-Id: I7246d4e68386e9f72559e8948c475e5916359922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1cec0e13e3e06081f3fdedf36d04597ef69ec962"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1cec0e13e3e06081f3fdedf36d04597ef69ec962"}]},"branch":"refs/heads/master"},"80db930895be9151f522642756cd458f280f263e":{"kind":"REWORK","_number":7,"created":"2024-07-26 12:09:12.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/85/924085/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/85/924085/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/924085/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/85/924085/7"}}},"commit":{"parents":[{"commit":"8416b72bff9d12f7baf097eb8efc9f7f5ddf8212","subject":"Re-join the strings after re-formatting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8416b72bff9d12f7baf097eb8efc9f7f5ddf8212"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-12 17:01:42.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-26 12:06:23.000000000","tz":120},"subject":"Enable mypy","message":"Enable mypy\n\nEnable mypy and fix/ignore found issues.\n\nWith mypy it is possible to declare/define optional or mandatory\nattributes in the base classes. Those are checked statically with mypy\nmaking existing validations and tests in the code unnecessary.\n\nChange-Id: I7246d4e68386e9f72559e8948c475e5916359922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/80db930895be9151f522642756cd458f280f263e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/80db930895be9151f522642756cd458f280f263e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
