)]}'
{"id":"openstack%2Fkeystone~924996","triplet_id":"openstack%2Fkeystone~master~I0b474143d07436d9b92cbe98e9caafe8318c2719","project":"openstack/keystone","branch":"master","topic":"renovate","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-08-27 21:26:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0b474143d07436d9b92cbe98e9caafe8318c2719","subject":"Add blackify commit to blame ignore","status":"MERGED","created":"2024-07-26 08:06:15.000000000","updated":"2024-08-27 21:29:19.000000000","submitted":"2024-08-27 21:26:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924996-renovate","meta_rev_id":"14f15441ac5fb16b5b1f912ffe6caf466ddd5c1f","_number":924996,"virtual_id_number":924996,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-27 21:26:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2024-08-27 19:37:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2024-08-21 15:26:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2024-08-27 19:37:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-26 08:22:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-26 09:33:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-21 15:26:28.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-08-27 19:37:02.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"2c74c60f06fc5087ead4b71f746b28861e485f35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-26 08:06:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"767368bbe065b8a69944f1f74754a6b60fb6f1ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 08:22:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8f10e81328c460f8126a8c7c3677d88\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ddfecb9d1af45acaf790cb07f733158 : SUCCESS in 15m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/63cb4ef0dff147fcb01c92a0cad0fe3b : SUCCESS in 10m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"00b39fc0acba37dae3615a8158c0cc7089ef6486","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 09:33:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b3d9aaa974c4f2880794cd676f6899a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c24e3c468c54f53becb404a2b6ade40 : SUCCESS in 15m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4557be9234b4be492731a450c2e4b42 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b336238f9d5494ca862df87f0de60e3 : SUCCESS in 11m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1745f54166d54d3390474ef6d180ad7e : SUCCESS in 9m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c7bd9569b97647cba4ce9a7aa10dfd1f : FAILURE in 13m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3dd4630df2e4e07ba540a00c34fe417 : SUCCESS in 12m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/90f32ffc4e1f4e6a9ce64be96804296f : SUCCESS in 1h 08m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3233006921f24c19b535f2185295f56d : SUCCESS in 1h 26m 37s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/811afe30ee164fc1b89b92c61507d2b0 : SUCCESS in 21m 57s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/272fabb95e88433195f0b8b8c320e256 : FAILURE in 24m 29s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5eb15a17c7043c4a8be6b98e7ca8445 : FAILURE in 29m 52s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/d8e85a639fd7461da9f585b9769499b7 : SUCCESS in 23m 39s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/67b73512cb364dc9b811bff0b276152a : SUCCESS in 12m 57s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9573fb65fb764ece88e821f24850a4e0 : FAILURE in 16m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5eed3bba48c492db58a22fb277a566c : SUCCESS in 1h 03m 33s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/aa0f80cd421a438b830d5388b3ad3872 : SUCCESS in 35m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29d0a29857b88b3b3814da4ba8886aa0a4048b04","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-08-21 15:26:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0515c920cb4f448eb03922c7cd5552943540f24a","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-08-27 19:37:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"502a0049418fdaf1d17dc15a623173d4ea74fc40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 19:37:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"855ed60d02d079492158152430a011dab6563e3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 21:26:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a58b719155034e26bb1877fc39fcdc5a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47e3b014cb5641e6b5841bc36e25c0f2 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3aadb0e262a44af4a9e4a0b9a30e9346 : SUCCESS in 13m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7245f8d2a1af4dd08f141965db9e6d19 : SUCCESS in 11m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfa5c90ebf1a4d1b82bfedb8f05cefb1 : SUCCESS in 13m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/f1cf3032a42a4e49bb934723c024fe4f : SUCCESS in 1h 07m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96f77f1f20e743da9f3dac0fda3e0b53 : SUCCESS in 1h 35m 20s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a92c2d27a4554e758a68dac9777e4633 : SUCCESS in 20m 06s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/44e0960d9b814d0aa8934c956027b6fd : SUCCESS in 21m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf76cf54102744548b84bfad1ab1c6a1 : SUCCESS in 50m 09s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/e9450531c1a44e978ba4f73d1f8eb85e : SUCCESS in 35m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc883fada6269d28a24ee5e7a3958eff1c17c57f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 21:26:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"14f15441ac5fb16b5b1f912ffe6caf466ddd5c1f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 21:29:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d852d25aa2e4872b9bf3909d9af45c4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0257641496c246aab176610471b553f9 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f66f8d4ade1e1b806b85711983255d245d9a0a8","revisions":{"5f66f8d4ade1e1b806b85711983255d245d9a0a8":{"kind":"REWORK","_number":1,"created":"2024-07-26 08:06:15.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/96/924996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/96/924996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/96/924996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/96/924996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/96/924996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/96/924996/1"}}},"commit":{"parents":[{"commit":"181dc7b03fec6b1868dc263d9804f6cdc34ddf1b","subject":"Merge \"Blackify the keystone code base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/181dc7b03fec6b1868dc263d9804f6cdc34ddf1b"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-26 08:05:15.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-07-26 08:05:15.000000000","tz":120},"subject":"Add blackify commit to blame ignore","message":"Add blackify commit to blame ignore\n\nIn order to keep git blame not overloaded with recent reformat changes\nadd the reformat commit to the ignore list.\n\nChange-Id: I0b474143d07436d9b92cbe98e9caafe8318c2719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5f66f8d4ade1e1b806b85711983255d245d9a0a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5f66f8d4ade1e1b806b85711983255d245d9a0a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
