)]}'
{"id":"openstack%2Fkeystone~926057","triplet_id":"openstack%2Fkeystone~master~I4ea5c179fcfd0a8e01ede1a4cdad23e67712e0db","project":"openstack/keystone","branch":"master","topic":"bug/2074018","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-08-21 11:32:48.000000000","reason":"Change was abandoned"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-08-13 11:11:44.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2024-08-21 11:32:48.000000000","reason":"Change was abandoned"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2024-08-21 11:32:48.000000000","reason":"Change was abandoned"},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2024-08-21 11:32:48.000000000","reason":"Change was abandoned"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-08-21 11:32:48.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4ea5c179fcfd0a8e01ede1a4cdad23e67712e0db","subject":"Revert \"Add keystone-manage reset_last_active command\"","status":"ABANDONED","created":"2024-08-09 12:48:03.000000000","updated":"2024-08-21 11:32:48.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"revert_of":924892,"meta_rev_id":"70e3659ee38d9d3b25b06d7735622d4b9893aead","_number":926057,"virtual_id_number":926057,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-08-09 14:26:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-08-09 14:25:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":-1,"date":"2024-08-09 14:03:53.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-09 12:48:03.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-08-09 12:48:03.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2024-08-09 12:48:03.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-08-09 12:48:03.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2024-08-09 12:48:03.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2024-08-09 12:48:03.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6e94e65340ccf1ee8605af40206bbc860eb7b772","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-08-09 12:48:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53fe9b59816d4e84fecf9e145e3e22781b289661","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-08-09 14:03:53.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d688f5cb091c7e14904b6e38a48a4386c3e51643","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 14:05:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f98c29e520814731860ba6ad9b1e1113\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bdac8dfc278a464ba94a1592973b502d : TIMED_OUT in 1h 00m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/53ecb0ccf2e7486096df5a0f864e9ae9 : TIMED_OUT in 1h 00m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b95c27cdeb74d22d32a6414ba7231618579d8e75","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-08-09 14:25:55.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0949854b4fdc5046dc0926c27dd784318237202","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 14:26:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64a815a1714b416faa006afdddb3dc93\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8df0618780a4439ca5992176fef791e2 : SUCCESS in 14m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acca0f3109ea4b3aa4dbe3d141c42f45 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47efcc11ccd2456389cdebe7ca40f2e1 : SUCCESS in 14m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c9814b24c532427eb31083279354982c : SUCCESS in 13m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/15e9d6058fc1434d8eb97a494aa35a1a : FAILURE in 12m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a023579f11cc4c2882cc1c18cd82ee05 : SUCCESS in 9m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/a22a4f03e5eb4c599f12a7b95ab6491e : SUCCESS in 1h 03m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1db9ce1caf074007be2e2443051f75e2 : SUCCESS in 1h 32m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01e58bd9d8b942b2820546c8388cd685 : SUCCESS in 11m 15s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d754bb65ab194d7ea361afb790d5959e : SUCCESS in 18m 26s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/96827286a738407ea658ba18f65b39d6 : SUCCESS in 37m 24s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcb030da80f242c98a7c73cfcd14f00d : FAILURE in 14m 17s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/4e23feb86c02491ca0f55f5a21b61ea5 : SUCCESS in 30m 01s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/a6c372383add4ebaaafa5929e5c0b18a : SUCCESS in 18m 47s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/a17d6b8fe09c41a1b511c96e09095e75 : FAILURE in 11m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e98468149cae495ba52c865063b4d506 : SUCCESS in 1h 11m 31s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/6e6744dc2a40465ca65c5a71cfc15d39 : FAILURE in 40m 27s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/f95f6f6fc0c64b00b616d17bbc9c431b : SUCCESS in 5m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a699e8aa287445b07d83c14f2817f8e00a3bce2f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-08-13 11:11:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"70e3659ee38d9d3b25b06d7735622d4b9893aead","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-08-21 11:32:48.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/oslo.serialization/+/926172 merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0783bfe070f73d3239c16fbac160b2d1b1bdd60e","revisions":{"0783bfe070f73d3239c16fbac160b2d1b1bdd60e":{"kind":"REWORK","_number":1,"created":"2024-08-09 12:48:03.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/57/926057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/57/926057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/57/926057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/57/926057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/57/926057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/57/926057/1"}}},"commit":{"parents":[{"commit":"e9513f8e4f25e1f20bc6fcab71d9177120000abf","subject":"Add keystone-manage reset_last_active command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e9513f8e4f25e1f20bc6fcab71d9177120000abf"}]}],"author":{"name":"yatin","email":"ykarel@redhat.com","date":"2024-08-09 12:48:03.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2024-08-09 12:48:03.000000000","tz":0},"subject":"Revert \"Add keystone-manage reset_last_active command\"","message":"Revert \"Add keystone-manage reset_last_active command\"\n\nThis reverts commit e9513f8e4f25e1f20bc6fcab71d9177120000abf.\n\nReason for revert: https://bugs.launchpad.net/neutron/+bug/2076430\n\nChange-Id: I4ea5c179fcfd0a8e01ede1a4cdad23e67712e0db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0783bfe070f73d3239c16fbac160b2d1b1bdd60e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0783bfe070f73d3239c16fbac160b2d1b1bdd60e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
