)]}'
{"id":"openstack%2Fkeystone~926351","triplet_id":"openstack%2Fkeystone~master~I613ae73f469321d2bbfaf74c4b3365c3365be9a8","project":"openstack/keystone","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-09-27 16:05:19.000000000","reason":"Change was submitted"},"35548":{"account":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"last_update":"2024-09-27 16:05:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I613ae73f469321d2bbfaf74c4b3365c3365be9a8","subject":"Prevent circlular imports in keystone.api","status":"MERGED","created":"2024-08-14 18:21:16.000000000","updated":"2024-09-27 16:07:46.000000000","submitted":"2024-09-27 16:05:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926351-openapi","meta_rev_id":"f56310dbc92361c07836860b023984ea9d9d5ebe","_number":926351,"virtual_id_number":926351,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-27 16:05:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"recommended":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"all":[{"value":1,"date":"2024-09-27 14:12:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":2,"date":"2024-09-06 14:49:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2024-09-20 14:18:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2024-09-20 14:18:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2024-09-11 13:49:53.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-14 19:43:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-06 14:49:41.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-09-06 14:53:27.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"CC"},{"updated":"2024-09-20 14:18:58.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-09-27 14:12:11.000000000","updated_by":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"reviewer":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"state":"REVIEWER"}],"messages":[{"id":"cfb06cf648c09a790384e035e90d1ae02bcad494","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-08-14 18:21:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cae3243ad88af2f04ddb49ff6a827b988ad045d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 19:43:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a8966e98da142e3b7f288e782894a22\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc8b30c48daf40bcb8bd29994741ddb5 : SUCCESS in 13m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c44f8ddc73e48c3a86e169f88520960 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4a90540b79b417ebf1767a4bf6d6030 : SUCCESS in 12m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a7a7cf3fdaab4df79e9436b93d0871e1 : SUCCESS in 9m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/753953a537474474b8a0166c61ae4781 : FAILURE in 14m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd68c5ad524e4465b3b37400cabc8f81 : SUCCESS in 10m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/5745e256c88f45cea70e66ad20fde98b : SUCCESS in 1h 03m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7061031265b545659b83018cc15f0597 : SUCCESS in 1h 21m 39s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e29d9b19f55a4eec904781cf2106a2f5 : SUCCESS in 30m 17s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/54c9275e906045479aa2df6fd96a6c39 : SUCCESS in 26m 12s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5990c1560a4c43db84d1d277d468ad47 : FAILURE in 25m 11s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/cd53016b503f4353b3b546ff90f68462 : SUCCESS in 29m 05s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/7d9c5193b443418c952022a8825439ba : SUCCESS in 9m 25s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/b6b4961df6b54a06bb6d784d16cb65db : FAILURE in 22m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebce5d90b9974116b285985ecb2eb417 : SUCCESS in 57m 32s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/8ccc25da363048e6ae977ba349b566b9 : SUCCESS in 41m 35s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/ff35c9da6e4b4dfb802b5b14e63ad62d : SUCCESS in 5m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f131141e910cfde2af574e4c8fe0b9047772ea36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 20:49:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2676035e588b46a88df46ba84c609279\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/203483e783474beeb48869ba03b72ba0 : TIMED_OUT in 1h 01m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/17ca7eb01db84495b6729aaa03d3037d : TIMED_OUT in 1h 00m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"697132f49a9eafb2e4c67123f72743e929f6ee48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-08-15 09:30:27.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"067b8132719f83b1f7872ee4e2153a27cda74e2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 10:34:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b80c64a698754b929f3dcd62b8d9a901\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/09406a393fa24cb68c832a51c6b42fa3 : TIMED_OUT in 1h 01m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/812bd380af0b488fb7e1352152a2c034 : TIMED_OUT in 1h 00m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1041da58c0dba1cb2c92556d0bab797985721b7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 10:49:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/910d493cc3964060aeaf67750f72eb5a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a832cc78bc6471c98eea30b162e1e62 : SUCCESS in 13m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99497d5dc0b84dd7ba72e2ab227ff53e : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdda41f47ce04f3797bcfdb56af38e10 : SUCCESS in 11m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a8535d67fef14a74b1d4a74b74a597ba : SUCCESS in 10m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/721fb399484846c48472d12ae1ea38aa : FAILURE in 11m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2b7dd76f4284023b287ca643da6896e : SUCCESS in 13m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/2bf9833d05714e3abac16a86f5d1983d : SUCCESS in 44m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2b6b5dd1c614e2ab80591bbbb506808 : SUCCESS in 1h 17m 36s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f31dcbf6197d4b839c9c1ad060cb1b0f : SUCCESS in 31m 42s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/4610c59633f0435b9f6b469170bded58 : SUCCESS in 40m 36s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb2bf9dcec6b423eb9ab6b3802c91b44 : FAILURE in 13m 50s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/cb532c3d671b4cc5b13e52d60fe0584d : SUCCESS in 19m 29s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/35b190a91fff4efea955c98f5c4a6ace : SUCCESS in 8m 43s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/d88234946ad549afa7ed41b7b22d7097 : FAILURE in 23m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4deedca7a25b40008ab07d0bbf474849 : SUCCESS in 48m 34s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/dc0ddf14fab44b05811464b0baa7a058 : SUCCESS in 35m 37s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/fdb60b775856488b8931e04f2f02189a : SUCCESS in 6m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"65b5edeae54bd8eceb4da4f2ac7ff8069695c87e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-08-15 12:21:21.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"68bdb00b7cc9e02d36d49733d4a385bfc2c21ba3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-08-15 13:23:56.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"afc9371997a9ca835a3bd9dd995b4ece4915a244","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 14:27:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aea4cb643ea84bd0b25d50e008f3fd83\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/68faa5dda73c4be2947f5d385f9d7043 : TIMED_OUT in 1h 01m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b74fd93b0479471abff3f66765fe8f09 : SUCCESS in 48m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"717534556debceaf5eb545cb1768b901f01844c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 14:46:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f1cf17e75384e8cbaff874368278528\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/999dc46b0032488aa76a5ed569580eaa : SUCCESS in 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cd4eaad4395485092bbee4bce76c9ba : FAILURE in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b16d12e3bb8d4b2ea60499608922647e : SUCCESS in 12m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/83ecaabaf1944235ac767ab272029ae8 : SUCCESS in 9m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6ce1c653cf9c4cfeaa1a2ee3e9e983cb : FAILURE in 9m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b133c8c0616b4242a161a52d55d2bf28 : SUCCESS in 12m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/5614cda5c3ec40b3abe711774abdad5b : SUCCESS in 58m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/33d0e47d37fa4ee98dd332231c00285a : SUCCESS in 1h 21m 09s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/899002a7b451486e8b0767657ebd5ec0 : SUCCESS in 19m 18s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/8504bdfb7ca44d2b9e825d82100c1b2a : SUCCESS in 34m 09s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e648941a3a24207a354d77373900455 : FAILURE in 13m 49s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/c5ee7600637f48b48a8d96412189f6f4 : SUCCESS in 30m 58s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/a11bce09a2eb48f0a54eb631dbe5ba7d : SUCCESS in 18m 20s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/97c835ad4c0f478299c8777c6f83ceff : FAILURE in 29m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cad0cf140e9649b9b204a91d48c2d63d : SUCCESS in 38m 54s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/625f7e0391b543cd89dc72ec1b213672 : SUCCESS in 35m 19s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/f4ac011ec0fd4ffb961e5b599ae868c9 : SUCCESS in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4ac9bf4f45f7654be2aed0c723e28b575f39d7ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"date":"2024-08-15 17:24:55.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"fed573d2df64401b8522a5e16576fe056fa45aa3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 18:27:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7433bb7a18564388957d3fd6bf59ef5f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2445f40895304118bd260141fc855dd9 : TIMED_OUT in 1h 00m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2f52add042184f0fa703d5923ca40da1 : SUCCESS in 33m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1e9cf323826b31aa57030a5ed9489ab7dc5c6065","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 18:33:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/624a0d74df60466a8325c9b46bb1cfc9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75e4b2453f6d45169cda06d3079027f1 : SUCCESS in 12m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61d285958c3349a2bfc1b6297f741efb : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d67f8f075ca40af9c81c624daf065cf : SUCCESS in 12m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/944b1db7d4aa4b7baa2f34d8e5b86612 : SUCCESS in 9m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/09bdb23a257e427c9459f441a8eaba93 : FAILURE in 12m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52be39b4a05d4300946548dfb694f0d3 : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/f66abdbc68094980b09302e76628e3e5 : SUCCESS in 46m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/369d3fb8d8204a9e80fcd7a061cbff12 : SUCCESS in 1h 07m 35s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b709d3b7d43f4c79ba2ad03278c38bf9 : SUCCESS in 20m 37s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/e1db62d14e814ce9a4ff265ff9d725ea : SUCCESS in 29m 35s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b383f9546896451da4d712ac6141e5e3 : FAILURE in 25m 24s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/0bdcc6d2760e48328a5eefbc6e4cdb9c : SUCCESS in 30m 23s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/ae90bcca698d40dd900f71e268dbf4f4 : SUCCESS in 21m 09s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9b5f77a4f3f94ccdbcbe67864c317036 : FAILURE in 17m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93599f6d707844ddba142afe9ae718f7 : SUCCESS in 52m 50s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/155fb001643d4868b75b5b222666cdba : SUCCESS in 35m 55s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/722eaafbb69e446589ab31de49c9e63b : SUCCESS in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fcf9e9782c0b831eb69101e5b1c197c0f4506726","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-08-16 14:14:04.000000000","message":"Topic openapi-patch8 removed","accounts_in_message":[],"_revision_number":5},{"id":"b804bea100d2d49dfb14b1c2cb21ba79c99a5a35","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-08-16 14:14:12.000000000","message":"Topic set to openapi","accounts_in_message":[],"_revision_number":5},{"id":"5f3b82bc63a4acffa09ca9157f457f7da65214a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"date":"2024-08-30 16:51:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"c6ead185c9d6e4c7d13a689a8e163eaba9541bea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 18:29:18.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f2b5e9b71604b7a844cf71f965ab90d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b61225e0492c4ee29a09c74f6508609c : SUCCESS in 16m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cc23d8742b74a778951cbfe1612081e : SUCCESS in 9m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad1fb9935ff0422a995e7053ba9d5ab0 : SUCCESS in 14m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6da95e59acc740ac8592963671ac08ee : SUCCESS in 11m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c233fdfbbee144828b996b17ec08326e : SUCCESS in 15m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c717de97e2234b7a944929dc26ef7279 : FAILURE in 8m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/6341036dacd040c18a91107ac294ba29 : SUCCESS in 1h 01m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ddb2384105994ebf979713540652c631 : SUCCESS in 1h 32m 32s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/628a42cc396248f68864141a2ff44055 : SUCCESS in 23m 53s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/f973ef6ef396414e9f48a5cf3000eefb : FAILURE in 30m 42s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c401c479ab34ee99e62cf8f778625cb : FAILURE in 17m 23s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/f6dbdb6509d34f408854f1eb4e7524c3 : SUCCESS in 27m 12s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/ef7c3dcc77b54c79a32acc5011b2a9a0 : SUCCESS in 12m 01s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/1c613bef2d3c413994986bc67e83d555 : FAILURE in 20m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a4bd32cd1934fada95fed733f31b0b9 : SUCCESS in 1h 01m 53s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/b03df75393294ee8a3f135481b7316eb : SUCCESS in 35m 53s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/5393949805cb4fa988e80ced3d2e53fe : RETRY_LIMIT in 5m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c0c3e17315595ab5d9a7d3879e82adb26ff0e2ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 18:45:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a10bf1c624a43f8bb9b9b68b204d110\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ebb9efbf855487f985f9b32826ddb73 : TIMED_OUT in 1h 01m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e85b319adf2d4f93adf30ec7813b11d5 : SUCCESS in 35m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"75f4ea4a8de871cae9a62daa6564db66e0db2f7e","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-09-06 14:47:19.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8b815c4efa81ed7a0c12c4a99a491e2c1f9ab598","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-09-06 14:49:41.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"96df54fa4d487eb7f6977564744764d84e622eb6","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-09-06 14:53:25.000000000","message":"Patch Set 6: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2ea958b1efaa5cde8097e951b62f2d2bb7d1c183","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-09-06 14:53:27.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b5ef52aa0ae80aae250ffaccb711e92bca8f388c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 16:02:49.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e900640db6d4fdb9400f23411b0817d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5b25ce7b2b78430bbe505ed64b40eac1 : SUCCESS in 49m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7b22b41e694e4596ab7c9d5b472f0a5a : SUCCESS in 40m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0680a298755cd3460ef361e2ba501e6d7e478538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 16:18:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcb52a08e10340829488f706f6e52ea7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf8daab352c542b5a99174ec5412d48a : SUCCESS in 15m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1edb39b044874a05a3c377610b43bded : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df69ade15a7f437c92478900315dc552 : SUCCESS in 11m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f6c871ced47c48dcbe5897084a3ab1ab : SUCCESS in 9m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/03fc9f6f88b1410092f713b6b3ff0f2b : SUCCESS in 12m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ab919e6ea5c489b8a7f2c881d207203 : SUCCESS in 12m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/51e51f070dfb4cee90540bc075593821 : SUCCESS in 52m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9caa8d8c05a0484b8f2ec0c6693d43aa : SUCCESS in 1h 26m 29s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ceda5c9d96f34b359d8034d87cdc20ad : SUCCESS in 28m 16s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/c23d64aa6edf4cba825cbc65fc845408 : SUCCESS in 37m 56s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8ebf55943184f018fb91b7ce7a65e96 : FAILURE in 22m 58s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/ae98ea180e824f3cb17d113fcc52ae5a : SUCCESS in 27m 14s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/592a378544994b86960e2e02a4151cec : SUCCESS in 18m 49s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/3f51f9a3336e4172a196031305931e68 : FAILURE in 20m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de37ddae117d4bb8afbe5dbba6342b92 : SUCCESS in 50m 42s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/a3a447e3f0a24eb5b7d883f4d5888aeb : FAILURE in 35m 17s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/74d7e12304c140929659bdaf712e29a4 : SUCCESS in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f0ca4de9295c8774bb6451019fa6a04a1bc4cb89","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-09-11 13:49:53.000000000","message":"Patch Set 6: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"aef11d11b42df63d688cc09141db67a253b9bf13","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-09-20 14:18:58.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"e1257d9c7b04e422cdbd7615cc7d6a53ae13ce7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 14:11:05.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4b16beffb8a329d3e4f5e7bc88cc6c396ac5878d","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2024-09-27 14:12:11.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"9e40277d726aca3281a45c5d0533d97dbbca683b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 16:05:19.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3aee17951734c12a233451d87b29fcc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a441d606fe424b6e9b85f407a2d91951 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32bf48548a6f43b492f6850763682321 : SUCCESS in 13m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/87ee43f5159d4081bcd8c013d6d5f54e : SUCCESS in 8m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dffe2d6b93cd431ba36ea37c4c375f3a : SUCCESS in 11m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/849c97ef27194a31bfa5d3bfe3a4e2c1 : SUCCESS in 38m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6da0c89d13b7417489d5cef762f1b583 : SUCCESS in 1h 30m 26s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/c1775d3b264049b0876b0de5a7f1609d : SUCCESS in 16m 19s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/7871e4ee4784432ea2f3d86600bf8e53 : SUCCESS in 31m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7e3f6f1211d4eb5a5437442a390a761 : SUCCESS in 1h 00m 53s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/d070326be7ac4f37b0ac746225181de4 : SUCCESS in 37m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ab2897ba2c7f9689521abdef856c9c88599200af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 16:05:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"f56310dbc92361c07836860b023984ea9d9d5ebe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 16:07:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec1c97aaaf174602861fdfd48435fe74\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d673b3f14e24418931ce403042eec14 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b00cb4197d603da948f8f43f9c5f0e55182d2c96","revisions":{"df0af66cdfb89b5c58eb0e84e085abf6cfa56720":{"kind":"REWORK","_number":1,"created":"2024-08-14 18:21:16.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/51/926351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/51/926351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/926351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/926351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/926351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/51/926351/1"}}},"commit":{"parents":[{"commit":"831d9098a25b4bd860ee8d22bbfab8f07b80bc78","subject":"Merge \"Add keystone-manage reset_last_active command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/831d9098a25b4bd860ee8d22bbfab8f07b80bc78"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-08-14 18:12:10.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-08-14 18:12:10.000000000","tz":120},"subject":"Prevent circlular imports in keystone.api","message":"Prevent circlular imports in keystone.api\n\nkeystone.api.__init__ is importing all submodules without real need. It\nis only used to make it easier to load all api blueprints.\n\nThis, however, leads to potential circular imports when inside\n`keystone.api` we import something outside `keystone.api` what is again\nimporting `keystone.api`. One example here started popping in the new\nvalidation framework which is located under `keystone.api` as well. We\ntry decorate api method with schema which imports parameter_types from\nthe `keystone.api.validation`. Since python executes all the `__init__`\nof the package a simple `from keystone.application_credential import\nschema` cause circular import (`schema` imports `api.validation` which\nimplicitly imports all submodules importing all other schemas including\nthe one we requested initially).\n\nOn the other side we also do not want that `from keystone.api import\nauth` would import all other submodules without anybody needing that.\n\nChange-Id: I613ae73f469321d2bbfaf74c4b3365c3365be9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/df0af66cdfb89b5c58eb0e84e085abf6cfa56720"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/df0af66cdfb89b5c58eb0e84e085abf6cfa56720"}]},"branch":"refs/heads/master"},"1e2a32ba866fc3270cc297138506e3228b997990":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-08-15 09:30:27.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/51/926351/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/51/926351/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/926351/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/926351/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/926351/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/51/926351/2"}}},"commit":{"parents":[{"commit":"9c936969023c8b23a12fed3c5379dd859835bab5","subject":"Add JsonSchema validation framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9c936969023c8b23a12fed3c5379dd859835bab5"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-08-14 18:12:10.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-08-15 08:25:30.000000000","tz":120},"subject":"Prevent circlular imports in keystone.api","message":"Prevent circlular imports in keystone.api\n\nkeystone.api.__init__ is importing all submodules without real need. It\nis only used to make it easier to load all api blueprints.\n\nThis, however, leads to potential circular imports when inside\n`keystone.api` we import something outside `keystone.api` what is again\nimporting `keystone.api`. One example here started popping in the new\nvalidation framework which is located under `keystone.api` as well. We\ntry decorate api method with schema which imports parameter_types from\nthe `keystone.api.validation`. Since python executes all the `__init__`\nof the package a simple `from keystone.application_credential import\nschema` cause circular import (`schema` imports `api.validation` which\nimplicitly imports all submodules importing all other schemas including\nthe one we requested initially).\n\nOn the other side we also do not want that `from keystone.api import\nauth` would import all other submodules without anybody needing that.\n\nChange-Id: I613ae73f469321d2bbfaf74c4b3365c3365be9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1e2a32ba866fc3270cc297138506e3228b997990"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1e2a32ba866fc3270cc297138506e3228b997990"}]},"branch":"refs/heads/master"},"80f15215cc02349cf290875aa0962bb7a27058c7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-08-15 12:21:21.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/51/926351/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/51/926351/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/926351/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/926351/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/926351/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/51/926351/3"}}},"commit":{"parents":[{"commit":"3a4a5b8fdd95f9ee7f1adb5fb724d13cce6d55af","subject":"Add JsonSchema validation framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3a4a5b8fdd95f9ee7f1adb5fb724d13cce6d55af"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-08-14 18:12:10.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-08-15 12:21:10.000000000","tz":120},"subject":"Prevent circlular imports in keystone.api","message":"Prevent circlular imports in keystone.api\n\nkeystone.api.__init__ is importing all submodules without real need. It\nis only used to make it easier to load all api blueprints.\n\nThis, however, leads to potential circular imports when inside\n`keystone.api` we import something outside `keystone.api` what is again\nimporting `keystone.api`. One example here started popping in the new\nvalidation framework which is located under `keystone.api` as well. We\ntry decorate api method with schema which imports parameter_types from\nthe `keystone.api.validation`. Since python executes all the `__init__`\nof the package a simple `from keystone.application_credential import\nschema` cause circular import (`schema` imports `api.validation` which\nimplicitly imports all submodules importing all other schemas including\nthe one we requested initially).\n\nOn the other side we also do not want that `from keystone.api import\nauth` would import all other submodules without anybody needing that.\n\nChange-Id: I613ae73f469321d2bbfaf74c4b3365c3365be9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/80f15215cc02349cf290875aa0962bb7a27058c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/80f15215cc02349cf290875aa0962bb7a27058c7"}]},"branch":"refs/heads/master"},"ea94a9abca4544a36891f60d554674455d454579":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-08-15 13:23:56.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/51/926351/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/51/926351/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/926351/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/926351/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/926351/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/51/926351/4"}}},"commit":{"parents":[{"commit":"a113dfae2d06c8aed02974ee313a30e85fb4ea7d","subject":"Add JsonSchema validation framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a113dfae2d06c8aed02974ee313a30e85fb4ea7d"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-08-14 18:12:10.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-08-15 13:23:48.000000000","tz":120},"subject":"Prevent circlular imports in keystone.api","message":"Prevent circlular imports in keystone.api\n\nkeystone.api.__init__ is importing all submodules without real need. It\nis only used to make it easier to load all api blueprints.\n\nThis, however, leads to potential circular imports when inside\n`keystone.api` we import something outside `keystone.api` what is again\nimporting `keystone.api`. One example here started popping in the new\nvalidation framework which is located under `keystone.api` as well. We\ntry decorate api method with schema which imports parameter_types from\nthe `keystone.api.validation`. Since python executes all the `__init__`\nof the package a simple `from keystone.application_credential import\nschema` cause circular import (`schema` imports `api.validation` which\nimplicitly imports all submodules importing all other schemas including\nthe one we requested initially).\n\nOn the other side we also do not want that `from keystone.api import\nauth` would import all other submodules without anybody needing that.\n\nChange-Id: I613ae73f469321d2bbfaf74c4b3365c3365be9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ea94a9abca4544a36891f60d554674455d454579"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ea94a9abca4544a36891f60d554674455d454579"}]},"branch":"refs/heads/master"},"19b1f846b7927491ee6df262027f34a4b2224c70":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-08-15 17:24:55.000000000","uploader":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"ref":"refs/changes/51/926351/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/51/926351/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/926351/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/926351/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/926351/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/51/926351/5"}}},"commit":{"parents":[{"commit":"45293c1f81c3d749d34a75c5d7895eb6198fcb13","subject":"Add JsonSchema validation framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/45293c1f81c3d749d34a75c5d7895eb6198fcb13"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-08-14 18:12:10.000000000","tz":120},"committer":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-08-15 17:12:58.000000000","tz":-420},"subject":"Prevent circlular imports in keystone.api","message":"Prevent circlular imports in keystone.api\n\nkeystone.api.__init__ is importing all submodules without real need. It\nis only used to make it easier to load all api blueprints.\n\nThis, however, leads to potential circular imports when inside\n`keystone.api` we import something outside `keystone.api` what is again\nimporting `keystone.api`. One example here started popping in the new\nvalidation framework which is located under `keystone.api` as well. We\ntry decorate api method with schema which imports parameter_types from\nthe `keystone.api.validation`. Since python executes all the `__init__`\nof the package a simple `from keystone.application_credential import\nschema` cause circular import (`schema` imports `api.validation` which\nimplicitly imports all submodules importing all other schemas including\nthe one we requested initially).\n\nOn the other side we also do not want that `from keystone.api import\nauth` would import all other submodules without anybody needing that.\n\nChange-Id: I613ae73f469321d2bbfaf74c4b3365c3365be9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/19b1f846b7927491ee6df262027f34a4b2224c70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/19b1f846b7927491ee6df262027f34a4b2224c70"}]},"branch":"refs/heads/master"},"b00cb4197d603da948f8f43f9c5f0e55182d2c96":{"kind":"REWORK","_number":6,"created":"2024-08-30 16:51:20.000000000","uploader":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"ref":"refs/changes/51/926351/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/51/926351/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/926351/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/926351/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/926351/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/51/926351/6"}}},"commit":{"parents":[{"commit":"a1d3a275b9152f08a6c0e21d80e0b45337ac18de","subject":"Add JsonSchema validation framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a1d3a275b9152f08a6c0e21d80e0b45337ac18de"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-08-14 18:12:10.000000000","tz":120},"committer":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-08-30 16:47:49.000000000","tz":-420},"subject":"Prevent circlular imports in keystone.api","message":"Prevent circlular imports in keystone.api\n\nkeystone.api.__init__ is importing all submodules without real need. It\nis only used to make it easier to load all api blueprints.\n\nThis, however, leads to potential circular imports when inside\n`keystone.api` we import something outside `keystone.api` what is again\nimporting `keystone.api`. One example here started popping in the new\nvalidation framework which is located under `keystone.api` as well. We\ntry decorate api method with schema which imports parameter_types from\nthe `keystone.api.validation`. Since python executes all the `__init__`\nof the package a simple `from keystone.application_credential import\nschema` cause circular import (`schema` imports `api.validation` which\nimplicitly imports all submodules importing all other schemas including\nthe one we requested initially).\n\nOn the other side we also do not want that `from keystone.api import\nauth` would import all other submodules without anybody needing that.\n\nChange-Id: I613ae73f469321d2bbfaf74c4b3365c3365be9a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b00cb4197d603da948f8f43f9c5f0e55182d2c96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b00cb4197d603da948f8f43f9c5f0e55182d2c96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
