)]}'
{"id":"openstack%2Fkeystone~926662","triplet_id":"openstack%2Fkeystone~stable%2F2023.2~I1b71fb3881dc041db01083fbb4f2592400096a31","project":"openstack/keystone","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-08-22 19:32:16.000000000","reason":"Change was submitted"},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-08-22 19:32:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1b71fb3881dc041db01083fbb4f2592400096a31","subject":"Add keystone-manage reset_last_active command","status":"MERGED","created":"2024-08-20 17:34:39.000000000","updated":"2024-08-22 19:34:07.000000000","submitted":"2024-08-22 19:32:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926662","meta_rev_id":"a3b60dcf8665bdbff32f6a4e5484b9c63dfcf400","_number":926662,"virtual_id_number":926662,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-22 19:32:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2024-08-22 17:52:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2024-08-21 07:37:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2024-08-22 17:52:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-20 18:39:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-20 19:06:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-21 07:37:46.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-08-21 07:40:03.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-08-22 17:52:00.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"347ee10dd1b9a50d97432a6ba1f64cc9991a3c5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-08-20 17:34:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"342bba3d02d30e72781acba736d9a4fde7ed3929","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 18:39:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c700d75a296144bdb3b07930f0a971f7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/06eedd8972934191a00175e0f7122e58 : SUCCESS in 45m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b30d4db8f59c46afad6fbd788ae5ae23 : SUCCESS in 42m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"78aa99f7381cef66c50c7aceb12ab91dafe79ba1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 19:06:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b078f279dbc94dc0a8039f764ca01fca\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6dcd233dd81442b91b91eff3e09cfc4 : SUCCESS in 16m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5c5631c04a945729ae89188c08f1618 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f114cc6b23c94817b060236b1e08fa8b : SUCCESS in 14m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10fa563466df42ce98999c6da0d0bf1f : SUCCESS in 12m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/010fda3ce4664c8995441e81a8b5ded9 : SUCCESS in 10m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b09141e48f6d47f5b2b489bc586d8ab1 : RETRY_LIMIT in 1m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5822ba38d6414f338974e2b58092cce7 : SUCCESS in 10m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/f73ec4c20b994cfaad587c336b233937 : SUCCESS in 40m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3884040ef802494ca48b89a87b0e203b : SUCCESS in 1h 29m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/beb804fcf9fa4b689b88aba0acb2c6fe : SUCCESS in 9m 53s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/7c0909f886c146298b769a2107e4aea7 : SUCCESS in 32m 12s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/d75f2ca858914770aeb4954e7816d4af : SUCCESS in 31m 59s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ca80b39f157445496cf81a7edf31581 : FAILURE in 24m 07s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/c5f3f1b6d85449299e7937194a830762 : SUCCESS in 19m 03s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/5f2b84e8b79a4a98b26faf35af01ecb3 : SUCCESS in 8m 22s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/b4db8c14d91a4dedb9b9b0cfdf0f2b2c : FAILURE in 26m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66f0cb9fe93648ae92a7bdcfcc86b94a : SUCCESS in 50m 06s","accounts_in_message":[],"_revision_number":1},{"id":"5ed2ade79c144827d607d08584971ef9475a8ef7","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-08-21 07:37:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e74fbdb0132c3c890faec05b32fa4f547f360b63","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-08-21 07:39:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a0319c1e258e4226fd925c2ab04b7d7afd9e1f6","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-08-22 17:52:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca331070c11cc074142341c60149dd90660c033b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 17:52:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c48df22bcccfc77b627af2ba76989792a0f38c83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 19:32:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f94e71bcfa0f4ee2a62907bbc34c557a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/594307f4031342a2be6a3aaabbaab2fc : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dadf3c08986c4292a2a43b8cd1b6294d : SUCCESS in 9m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75dfd892f8d14cb7912b450c3f188c1e : SUCCESS in 9m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac71b3b0f35841b4bde5442575660ea1 : SUCCESS in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81b66343dbbf4da6979494f664ddd73f : SUCCESS in 11m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/a96e8507f08e4eeb826fee2b18baecf7 : SUCCESS in 1h 03m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd13b1c067c346f192316e10d42b31b3 : SUCCESS in 1h 39m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2985e31957884801a3c558646c9741c6 : SUCCESS in 8m 10s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/66fcf6a4592c491e88b53d8670fc6b15 : SUCCESS in 30m 08s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/117cb1e731234a449c45d8e04e84307e : SUCCESS in 29m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5b82b2d0f584eb8a92e9f1a85ddef83 : SUCCESS in 57m 09s","accounts_in_message":[],"_revision_number":1},{"id":"a53f7498bbb2359bfdd12fd8746621074c5c506d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 19:32:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a3b60dcf8665bdbff32f6a4e5484b9c63dfcf400","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 19:34:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09012c991b314ef8a69be4102abc3610\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47b631ff37cb4984b88e76661e606c51 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d36840c7af74073b2532cc2c5b5ec90 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1829ce0202cbe11602fafd053cd1cd3eb0f996b0","revisions":{"1829ce0202cbe11602fafd053cd1cd3eb0f996b0":{"kind":"REWORK","_number":1,"created":"2024-08-20 17:34:39.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/62/926662/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/62/926662/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/62/926662/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/62/926662/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/62/926662/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/62/926662/1"}}},"commit":{"parents":[{"commit":"fee954e9b48fa7a57bdf3ef98e50f743fda969f5","subject":"Remove reference to devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fee954e9b48fa7a57bdf3ef98e50f743fda969f5"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-07-19 21:10:11.000000000","tz":-240},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-08-20 17:34:29.000000000","tz":-240},"subject":"Add keystone-manage reset_last_active command","message":"Add keystone-manage reset_last_active command\n\nThis patch adds the `reset_last_active` subcommand to the\n`keystone-manage` command line tool.\n\nThis subcommand will update every user in the database that has a null\nvalue in the `last_active_at` property to the current server time. This\nis necessary to prevent user lockout in deployments that have been\nrunning for a long time without `disable_user_account_days_inactive` and\nlater decide to turn it on.\n\nThis patch also includes a change to the logic that sets\n`last_active_at` to fix the root issue of the lockout.\n\nCloses-Bug: 2074018\nChange-Id: I1b71fb3881dc041db01083fbb4f2592400096a31\n(cherry picked from commit e9513f8e4f25e1f20bc6fcab71d9177120000abf)\n(cherry picked from commit 62b3c9260b79b9596545debbd7e4fcd3b4eab972)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1829ce0202cbe11602fafd053cd1cd3eb0f996b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1829ce0202cbe11602fafd053cd1cd3eb0f996b0"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
