)]}'
{"id":"openstack%2Fkeystone~926776","triplet_id":"openstack%2Fkeystone~master~Ia4b4dc14bc4163463ef4271851b117e36baae0ec","project":"openstack/keystone","branch":"master","topic":"use_async_for_reader","attention_set":{"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2024-08-21 15:56:11.000000000","reason":"A robot voted negatively on a label"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-08-21 14:25:40.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ia4b4dc14bc4163463ef4271851b117e36baae0ec","subject":"enable use of reader session, isolating writer session","status":"NEW","created":"2024-08-21 14:01:47.000000000","updated":"2024-08-22 02:35:30.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c2eb45f6cdd540d49ca070addcd12ca6afd65c27","_number":926776,"virtual_id_number":926776,"owner":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-08-21 15:56:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-21 14:25:40.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-08-21 15:56:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c644c94676e8bb5383202b80947fed2cbac8152e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-08-21 14:01:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b67e91542139ea45e1ec558a4879e6b11c281e9","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-08-21 14:06:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f440682a05faa0c7d52cb7c68db30e6063cb8b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 15:56:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bfdd37f88aab4bd2bb47032a46729c7e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae7be289054e49de8a595f96f7a5d8cf : SUCCESS in 14m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/203e2ebaa7d44aedb555afe906144afc : FAILURE in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6e30e8f4e3347628fbda77455aa9b4a : SUCCESS in 12m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d73f51f7fb64343b9394e901c5c9436 : SUCCESS in 11m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8f89d744350c469fbfdeb46bcaad8f8f : SUCCESS in 11m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b969f5d5b93a4fa79138c0526a731cc3 : SUCCESS in 11m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/23e2aeeb13d1469b9c94912893fdacf8 : SUCCESS in 53m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8bb3bf415d7a4e74abc3684c2cd84b3a : SUCCESS in 1h 44m 51s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/aa42f91b664c470baa435757f1832f3f : SUCCESS in 28m 29s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/7ce84f020119409f9cd713edd809efb2 : SUCCESS in 31m 08s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3969c773a2614c4d834d3281c2441547 : FAILURE in 29m 32s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/afb3339ff8004c6b83a0062a95c4b26d : SUCCESS in 30m 54s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/752acbe4abe74a6dbd0fac08fd58e098 : FAILURE in 19m 51s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/1bcc0222818d40ac9056c847ff9a42c0 : FAILURE in 22m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/126b1c5b17594549a274573b33a8432f : SUCCESS in 56m 57s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/168a8e8f4aed4d3184e740eed6f65f2e : SUCCESS in 48m 32s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/e31d98db1f3e454c9eec8dd60aa56272 : SUCCESS in 6m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c2eb45f6cdd540d49ca070addcd12ca6afd65c27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 02:35:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaf4006d16ec47de851fcc273cce1bbe\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7202f2c1743a4291b36c9960b35fb53a : TIMED_OUT in 1h 01m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a2f95853b0b94cb0a00aadb14c4e2261 : TIMED_OUT in 1h 01m 00s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"05c66e3727537d4ee222e85fd705f1fa6c2118a8","revisions":{"05c66e3727537d4ee222e85fd705f1fa6c2118a8":{"kind":"REWORK","_number":1,"created":"2024-08-21 14:01:47.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/76/926776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/76/926776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/926776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/926776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/926776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/76/926776/1"}}},"commit":{"parents":[{"commit":"831d9098a25b4bd860ee8d22bbfab8f07b80bc78","subject":"Merge \"Add keystone-manage reset_last_active command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/831d9098a25b4bd860ee8d22bbfab8f07b80bc78"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-08-21 14:00:07.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-08-21 14:00:07.000000000","tz":-240},"subject":"enable use of reader session, isolating writer session","message":"enable use of reader session, isolating writer session\n\noslo_db enginefacade only uses the \"reader\" session if\nall of the following are true:\n\n1. the \".reader\" modifier is selected\n2. the \".async_\" modifier is selected, or synchronous_reader is False\n\nkeystone was doing 1. but neither of 2., so set that up.\n\nadditionally disable the legacy \"test connection\" feature that is\nnot useful, thereby preventing the writer session from being used\n\nChange-Id: Ia4b4dc14bc4163463ef4271851b117e36baae0ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/05c66e3727537d4ee222e85fd705f1fa6c2118a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/05c66e3727537d4ee222e85fd705f1fa6c2118a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
