)]}'
{"id":"openstack%2Fkeystone~926936","triplet_id":"openstack%2Fkeystone~master~I00cc5a3f35747411aaeac4ef41da9c3793da5530","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-08-27 21:26:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I00cc5a3f35747411aaeac4ef41da9c3793da5530","subject":"Update hacking to latest version","status":"MERGED","created":"2024-08-22 17:01:37.000000000","updated":"2024-08-27 21:28:17.000000000","submitted":"2024-08-27 21:26:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926936","meta_rev_id":"9752748301f3c8a52a2249e2461ab0105f4ddb68","_number":926936,"virtual_id_number":926936,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-27 21:26:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2024-08-27 19:35:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2024-08-26 14:38:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2024-08-27 19:35:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-22 18:39:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-26 14:38:27.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-08-27 19:35:47.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"985736e0b970a5fa8bf0f64833f457784f29b362","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-08-22 17:01:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9630a5c50e5eff55fba2af27e90dc1cd74dafff1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 18:39:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2ffc284a382444f802e17477b1f7b2e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d4d8284b7d848eeac91c1d5fb66ddf6 : SUCCESS in 14m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/894bdd3bf8344140a6dcfd33676ea690 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8670ee8913bd407eb638623b3893ce84 : SUCCESS in 15m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/007e71d88ddf4779938ccdf1644355e1 : SUCCESS in 10m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f177378d870a4f7ead026c29d8fb39d6 : SUCCESS in 12m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65f78ad888ab4c82ad783f6fbb265581 : SUCCESS in 10m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/1503ecaea17e4cd483a1458ba26aed0d : SUCCESS in 56m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25b2d7bfcaf04f0c822aa330216f6fc5 : SUCCESS in 1h 35m 33s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/af1b87e427f443639c9fd2a1562b0b6e : SUCCESS in 21m 28s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/373ffe09adba42beb60915c3ae7f3be0 : SUCCESS in 42m 02s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c796e2ee180341a3879c2d5a32db7a62 : FAILURE in 27m 17s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/efd44c9457214fc18ebf63c68717da7d : SUCCESS in 27m 26s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/f9f048525aa241e281e54d1fb8301b45 : SUCCESS in 22m 38s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/7b096139077c4a51bf78241bb5bcdb92 : FAILURE in 12m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9b0a0a4f5834345911d468808075088 : SUCCESS in 39m 39s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/2216336206f5490788abe84fe0b56b4e : SUCCESS in 35m 32s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/eec5efc8c08c4c11b7b02045f0e77514 : RETRY_LIMIT in 5m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6350267704494d5115ecb47c5937b2acd15874f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 20:24:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6548fc5f4c8d4f2ebe34c575f34e2edd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0d82ae2824b348b384e367e3ff622a55 : SUCCESS in 42m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/127a4c76e178472397581a72504305c9 : TIMED_OUT in 1h 01m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fec2d89e180855f0bac67babaa226927df527db9","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-08-26 14:38:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5ab4c877c9e7c0f7c8c52321c6b430de2fe6ae41","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-08-27 19:35:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2149797edf537d7128bc4500f11ad4ad6c425ec6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 19:36:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12db5ee5809322143dff5a6f6a1778f57a5947fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 21:26:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/627a9fa1f825451894c1a85b653ab87a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e70552fa5a5c4b479275616d6979b7e9 : SUCCESS in 8m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4912b3a5f95483fb207e3bd83e6edc9 : SUCCESS in 13m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ca32aa3dbf474eed9cddfe081299d1e5 : SUCCESS in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4ad9229419a46c198219a455b6d699b : SUCCESS in 14m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/dcff1f5811f04a0b8c7f98f3d36e9348 : SUCCESS in 54m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6369385afe954ce0a75f05a79b51f2a1 : SUCCESS in 1h 40m 33s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/4bc77dcf51a949e2ae4e24e052ebf585 : SUCCESS in 19m 30s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/22690ab14f4a43898c57dd7d42260b23 : SUCCESS in 24m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/509496ba96b54bf9b584fad609aae578 : SUCCESS in 44m 00s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/1d8700bf5d2d491fa192771e678f33e6 : SUCCESS in 30m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d774a7a303314bc8e9b7e3ef00cd66d0e57a75d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 21:26:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9752748301f3c8a52a2249e2461ab0105f4ddb68","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 21:28:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ce2e8b33d34409a93c1298463271d40\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be07fc9529f441b2a5782ee7f9ef59b6 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8c3113b5131a1769bce7cfa91d41707da094b829","revisions":{"8c3113b5131a1769bce7cfa91d41707da094b829":{"kind":"REWORK","_number":1,"created":"2024-08-22 17:01:37.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/36/926936/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/36/926936/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/926936/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/926936/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/926936/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/36/926936/1"}}},"commit":{"parents":[{"commit":"0d293d5c7e68f2397c5369103f470399d00c3ed6","subject":"Enable hacking check in pre-commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0d293d5c7e68f2397c5369103f470399d00c3ed6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-08-22 17:00:55.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-08-22 17:00:55.000000000","tz":120},"subject":"Update hacking to latest version","message":"Update hacking to latest version\n\nChange-Id: I00cc5a3f35747411aaeac4ef41da9c3793da5530\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8c3113b5131a1769bce7cfa91d41707da094b829"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8c3113b5131a1769bce7cfa91d41707da094b829"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
