)]}'
{"id":"openstack%2Fkeystone~927523","triplet_id":"openstack%2Fkeystone~master~Iccef6011ff63f78d43a02fd2ccebbf6ac2522583","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-02-21 14:20:15.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"35548":{"account":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"last_update":"2025-02-21 14:20:40.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-02-21 14:20:11.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Iccef6011ff63f78d43a02fd2ccebbf6ac2522583","subject":"Add JSON Schema to `services` and validation decorators to services resource.","status":"MERGED","created":"2024-08-29 15:37:37.000000000","updated":"2025-02-21 16:33:53.000000000","submitted":"2025-02-21 16:32:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927523","meta_rev_id":"42831eaa063fe6491bf9f9c8e5e297505a792e5c","_number":927523,"virtual_id_number":927523,"owner":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-21 16:32:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2025-02-17 11:04:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-21 14:09:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":1,"date":"2025-02-21 14:20:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-21 14:09:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-29 16:39:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-17 10:41:50.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"CC"},{"updated":"2024-09-17 10:41:50.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2025-02-17 11:04:34.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2025-02-21 14:09:54.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-02-21 14:20:09.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"3304b1ffd61fa246d44363667f1370eb77c87b37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"date":"2024-08-29 15:37:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"149a2946c3940d087aac27f0acf070ef160cbeef","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"date":"2024-08-29 15:37:47.000000000","message":"Topic openapi-patch8 removed","accounts_in_message":[],"_revision_number":1},{"id":"4956c92105319d531017b7e6f2984b5878c7cb62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 16:39:08.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1eba804d40de40e68776f7225050b1af\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac7fa9b375e24b5f9e16fca9be87f785 : FAILURE in 14m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2532484a8aef40ca86fd942fca0761f0 : FAILURE in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d5b309428e541458881f4b77e4d3f0d : FAILURE in 14m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f617ff98a9f4531851ea25cf93bf0c1 : FAILURE in 12m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a2fee32dff1f415aace375afb50aecff : FAILURE in 14m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75afc1f891ad4fd3be3c1b22e39b6fb3 : FAILURE in 10m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/d5dd2de428d14b55a1d39b5906c47965 : FAILURE in 59m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/51c264d43ac643aa8b404f3f67faf155 : FAILURE in 18m 27s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e0cbce108dd24540914d0445a356255c : FAILURE in 16m 43s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/bfcde74eeb7a4824a9405bad8b983dea : FAILURE in 25m 45s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7eec408536c249f292e929ec10fefd36 : FAILURE in 15m 01s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/aaebf1a9522f443386d44c958b5fbf44 : FAILURE in 15m 17s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/f445a707c4d84ed2be42579504e0bf28 : FAILURE in 9m 34s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/340d5b741eb243c9b96a5e72b1967a61 : FAILURE in 12m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/868ce6fa5dd84401afb1e530200e5f22 : FAILURE in 16m 38s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/54f34bb6fc8f47aa8aefd712a41d7296 : FAILURE in 12m 46s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/90527f43d6f3403b9983abf1b3021abc : RETRY_LIMIT in 5m 46s (non-voting)\n\nWarning:\n  Comments left for invalid file keystone/api/__init__.py\n  Comments left for invalid file keystone/api/validation/__init__.py\n  Comments left for invalid file keystone/server/flask/application.py","accounts_in_message":[],"_revision_number":1},{"id":"d297c37106ef6deda9bfb5687f7bb906cc571730","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 16:40:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb332ad360d24cb28c44d011ff98de57\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e14d16aa02b4f1db4b81f55177ef0ea : TIMED_OUT in 1h 00m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b550e81023ab44959a2ff5aaff34ecb3 : FAILURE in 45m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f46684142aa7eab9748cc6d3b99a642772e80d42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"date":"2024-08-29 17:24:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"de214839d02b8635f28c27618df70b83080df38d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 18:23:28.000000000","message":"Patch Set 2: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af0fd1008b6c4cefbd140d4910969f7f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4e5249dc1fd40119986386510a24ea8 : FAILURE in 13m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5ad247b1f66414fbfd56daa627fa8f5 : FAILURE in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4de6267cd064d1496f44bb5154f3a68 : FAILURE in 11m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f3fd960879148d9b97d2fcef09aa80f : FAILURE in 9m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9eb9c6a1744e47548cd64c2549e6d972 : FAILURE in 12m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31da73b70a214ffcb7f31feec9caf7c2 : FAILURE in 7m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/a94d83be03fa47f99df11b761e66be8a : FAILURE in 58m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eea6c8e28dbc4705b8bfa43a88791e38 : FAILURE in 24m 14s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/7ce37d56a6154fd2a8914938f3c6b397 : FAILURE in 30m 49s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/555542d90dc44199acf002efe3ccad73 : FAILURE in 35m 39s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0af16613914a4c8ba61bf6d4642ae655 : FAILURE in 26m 20s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/d8fc438d360a4c7eb9b38a40bf39b8a8 : FAILURE in 26m 38s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/27db1dc7b25641d89cde2abdd61199f6 : FAILURE in 19m 10s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/38bf6c3d33524a3a9edcbdc73c318a0b : FAILURE in 22m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2234d5262e364896a899df7effb611f0 : FAILURE in 24m 12s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/62ea3a885c204b54bd4dd5a37a4f7029 : FAILURE in 24m 27s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/f660da36de95471ba4b81b8f3989eb96 : RETRY_LIMIT in 5m 22s (non-voting)\n\nWarning:\n  Comments left for invalid file keystone/api/__init__.py\n  Comments left for invalid file keystone/api/validation/__init__.py\n  Comments left for invalid file keystone/server/flask/application.py","accounts_in_message":[],"_revision_number":2},{"id":"e3e124c67a1c180288be55f6c2e0b7e90cef9077","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 19:06:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0fd547d9c934f9fbffb55dbf03a201c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a1a05d2c575a4e9ea3b0a8e9e3d0b648 : TIMED_OUT in 1h 00m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/db06c927fea04b1d8b0ac441a4025c03 : FAILURE in 43m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"50740ad72510a29a219e0eef503a6f2d29a925d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"date":"2024-08-29 20:30:26.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"00a38c3bb4715b2e57bccb1cb684ce6b4c17cfbc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 21:13:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90c1043b3c6a433d878bd824d4b532a6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/48ea0b3c61df4f21aeebb97f37d06ca0 : FAILURE in 41m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6b89f8e1002541d6ab5fdc276c605402 : FAILURE in 31m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fa502dc3a5460e1420fb03c7983cedf210ae62a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 21:27:47.000000000","message":"Patch Set 3: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a00b2dc64f3b49a3af3ece3f4a4b8a80\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2748b34cff940dcbe3c62ecfe92b967 : FAILURE in 13m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4490ca99b910468fa630dd7c474d77bf : FAILURE in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddd9507e9bc74a08bb45ba84f101cb68 : FAILURE in 11m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70beec806c974195ab28f8dbe9b20ca8 : FAILURE in 11m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e9192d14fbe74a08a90a322626fe1cf2 : FAILURE in 12m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb5197e5b3424bf1bd4ef18e6af1acf5 : FAILURE in 6m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/48bc00a234b1414aa93f34ef3291d5c4 : FAILURE in 54m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b87c8dc92e244039ce8448747d586ad : FAILURE in 14m 33s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/aa7ccced7c1b4844b71402df3e00b6a5 : FAILURE in 26m 25s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/d74a4962081a43b4813907dc899e4f4a : FAILURE in 37m 49s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff43e93af3984b6d97477bba7d7141c8 : FAILURE in 25m 18s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/bdd837b880044ec39c78774ac84b97ab : FAILURE in 20m 18s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d45392a1ec3b47f8a2622d7f32aaa9c7 : FAILURE in 20m 20s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c377b4488a934725a01ec8b64e9a91ad : FAILURE in 22m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90ca875db88c4a61900d5339affe66b2 : FAILURE in 12m 57s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/80375e352e7e4909b455b1a1a5f693bf : FAILURE in 12m 20s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/7b838a9c96974825b75b55755c61cf2c : RETRY_LIMIT in 5m 12s (non-voting)\n\nWarning:\n  Comments left for invalid file keystone/api/__init__.py\n  Comments left for invalid file keystone/api/validation/__init__.py\n  Comments left for invalid file keystone/server/flask/application.py","accounts_in_message":[],"_revision_number":3},{"id":"e47f728a2623df9c558873b3afb6e460004e100b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"date":"2024-08-30 16:03:07.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"163292cc746964ddb9667a36bc3d45acd7d9ab19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"date":"2024-08-30 16:51:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7b040be97b549ff6e791c6569d83dcb9723e9013","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 18:30:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ee38c4a6ea64434999b47301073a005\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bab3ae897dda4a978043c34cc24916a1 : FAILURE in 12m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9285c4117fb747e3962964b0e36f00aa : SUCCESS in 7m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb6ba739b1c9434d91203046d973432a : FAILURE in 12m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7a92fbcb59b5483fbac3bb41c9d2d943 : FAILURE in 10m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6ed11698e0e043f0b3d6690df1fe6ada : FAILURE in 13m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d21af1a36d5f4114b38fec0ff5b0aa90 : FAILURE in 8m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/deba3584dc9940bebc42a56a28380c7f : SUCCESS in 57m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb8b4c2e4fe34f3abdcbd46db3503150 : SUCCESS in 1h 31m 37s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/766524deb10640a5807fba400ce8f786 : SUCCESS in 28m 55s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/6b4a83553f514dc2991afd92b3659458 : FAILURE in 28m 09s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55220212731c4884b730f26912ee884f : FAILURE in 16m 45s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/9a820373a5454db0a6bd0109ff4bf584 : SUCCESS in 26m 37s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/2e48efdc5ee549d1bced17b056c79f5b : SUCCESS in 22m 09s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/91fa99f029d741509208ad1fbacf3cc3 : FAILURE in 13m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3999a7b309aa4e6092bc863cfb105af7 : SUCCESS in 52m 26s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/287d4650b83e4eacb694c1dd5195eacc : SUCCESS in 35m 20s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/9fa9fb78567b4f4c98358680db7d2293 : RETRY_LIMIT in 7m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0f03faec1f920771d6c310174788a76d20381f77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 19:26:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/514b2744300a49d09f74a18c4b83a67d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ee3a979faa34ae8b7335f462e8f79ab : FAILURE in 1h 00m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/479d986f7edf4288badae7cf66632b67 : FAILURE in 41m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2e59233a0c366b155eef331823de4ee6a7802785","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"date":"2024-09-03 16:11:08.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"ae074376a9cba066745836f8534413ce30d5dc27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 17:14:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13898bd7fbca4c029fae4e3522bf73f2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/358fc820556b4288b2462848f76be476 : TIMED_OUT in 1h 01m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/caf15f7d2d8b410087801b5cd5309412 : TIMED_OUT in 1h 01m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"26e0e802a56ca7feacb29e372bb1ccd4ec9710fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 17:50:23.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b479725fd2844558b8cafefe69ef235\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03dadf1342a341ac9ed3cbd6c1710bab : FAILURE in 14m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6285e7f46cbd4d15af58d9d458715f0d : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df084f05158a42ce95d52702ed1f4911 : FAILURE in 12m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b3ed5e97074d4f4dba6b5fc2423feb96 : FAILURE in 12m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1c62b1ac2be44766b8db3e0799d08050 : FAILURE in 14m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a776ab1cb9c4a0b861bf5d467ef5934 : SUCCESS in 13m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/9edd77a3bfce410799d01801cb14af6c : SUCCESS in 1h 07m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2cbf391beb0341edbc30f24af6989e93 : SUCCESS in 1h 38m 20s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/54d164f0d6864334845c4ea324c58a8d : SUCCESS in 31m 02s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/f096431002574f1cb4cf5b0df0185582 : FAILURE in 21m 19s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e21d37932a704c5db9461636bdabb5dd : FAILURE in 23m 51s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/5a18544b2a284f76ac81f11402315999 : SUCCESS in 28m 11s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/533d4b5c65b44247b08dbcc776edd09e : SUCCESS in 19m 05s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/2e7e81e693a548d3bcd44edfcdff9303 : FAILURE in 15m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57cdb677064a405eb5cdfa08394c3eb1 : SUCCESS in 49m 50s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/8ddba16815184d3fa3c525634947ceed : FAILURE in 36m 54s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/185046079db74b27a75c65d6cdf0c35e : RETRY_LIMIT in 5m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"206fa776690f843bc8265c7fb4d86d57e42cddf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"date":"2024-09-09 17:32:43.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"b44672bcee305349052128589ed57f7b6ab86ca5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 19:36:46.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca0258a72ecf430f89188ebea653eb50\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0c78bfe2f499480c97378abb9495ec51 : FAILURE in 54m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dc6a35fd5d094efda8391d46545555db : FAILURE in 36m 43s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f16604c1eaaa55c80e018f32019dd2563678b233","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 19:53:36.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/198c883e99fe4010a49a41527da38852\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4ceaa01744b48a38b3a66722193a2f1 : FAILURE in 14m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ebbccbc00b04b7faf5c3f1c43a90ddf : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b69b91e957b9404980428eb4c39eba2e : FAILURE in 13m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c3dddb6f7e0e4cd0b97aa785a590585e : FAILURE in 10m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7a7d01f643a94ff3a32f449308b3d547 : FAILURE in 15m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/600413a2741e4f05baba722603159358 : SUCCESS in 11m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/59a4871951e64729a36f3f8ab4cb4122 : SUCCESS in 1h 01m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/770dcc6813ff44759970e5ea9f9f5ab0 : SUCCESS in 1h 39m 04s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d839c6f59a5b4149b7ddf301acda89b9 : SUCCESS in 20m 05s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/82a98f6b92c945d196f588a64794ed07 : SUCCESS in 30m 21s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aef797ec058945c48f3f9c6f15fee86d : FAILURE in 18m 28s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/6960a8e468de4e26842d5d2b5b7a30a0 : SUCCESS in 32m 43s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/015cfc49d4254ff49db89bb6dcf4817c : SUCCESS in 10m 14s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/4b035e9856e44811a5004bbf99c01f01 : FAILURE in 21m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4dc0fd4a95d4648b7d7ae1b07288e57 : SUCCESS in 50m 49s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/995ecda80f8e45e4ae8d1e5a4ecbf41e : SUCCESS in 35m 23s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/14145dd76f5341e298fb02845e7e283c : SUCCESS in 7m 14s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"494d1fbf9fb49eb2b9335cbddca52c8c540c3632","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"date":"2024-09-10 15:50:04.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"592f2619321412dc51e5db36c062e724bdb6b774","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 17:24:37.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a75fc4bfbbfe49de8be74fa5a067aa47\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c2589a5ba15d43af8a1b05de328a9919 : TIMED_OUT in 1h 00m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7df0cbe41dfb4453b346ef60f64ba399 : TIMED_OUT in 1h 01m 16s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ad646f353a4ec45c5a94588043436b7511e2c706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 17:38:14.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fe8954b818842ca8b4cea3a18f68ba3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77ae09a8a6f74101b97b6e70374cd00b : SUCCESS in 17m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baeeb14496eb42b1965084a3ebc1172e : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6bdcb7e83c74136a0a20b906790d2d1 : SUCCESS in 10m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e6afb2d618024f3ba1d2993f4928019f : SUCCESS in 11m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d83fe5ccc7ec45b894b488a80a3c45a6 : SUCCESS in 11m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e62c0ae8c04b4a688d3ce5d9813b1c14 : SUCCESS in 12m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/ef0a77b220ea47a8b5d5c0cfb21d0b7b : SUCCESS in 55m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/860a679e6c934cf2b422a469cad33586 : SUCCESS in 1h 45m 45s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/571b5b3d36ad4d3fbffbc42cef58fe20 : SUCCESS in 25m 54s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/8718c45a91704d3a864e5f9ee8cbb9c0 : SUCCESS in 30m 06s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae9d09419b624c28b80cd3e625604290 : FAILURE in 17m 23s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/e6f5b716dd0446c18b747e71329fcdee : SUCCESS in 22m 21s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c234bb100ed8485f96668cad5c61e00b : SUCCESS in 8m 28s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/73c92bdbf9ee44f3a11a51eeae8b3251 : FAILURE in 13m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bc941b0de7b4a3fbf9b22299afa66ba : SUCCESS in 44m 26s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/ae4535a518aa497e80f61e9ffbfd75a7 : SUCCESS in 40m 14s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/e1c6aaaf7d40452baf677852dc2dfcf4 : SUCCESS in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"28b26bee465e566c6580855d6f5aad35c83e98da","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-17 10:41:50.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"ea6e11e7ea9f6f9e15c816eaa94117419aceeebb","author":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"date":"2024-09-17 16:21:45.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"ed78142ee26bb4df22f4ea5c79290f90188f7360","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"date":"2024-09-17 16:29:31.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"664cd6810e17f4d36c339f9eda4427691a170156","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-09-17 16:36:19.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"46711a165e2952380d85b39400e6318aa8ac0d24","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-17 17:39:05.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d68ca3c3c3b21b28a2669a5ccc3e2ece1d78a482","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 17:40:13.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73add0792e164eda87a2198fdd63a8e5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b91d8460ae246daa47af9ffb51841c4 : SUCCESS in 13m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5ca738250184bd2a8030479c066c6e7 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b137b62a7c443bdb8a589204ed629f1 : SUCCESS in 15m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7db7a114d9664a71a1407bc37121c7c6 : SUCCESS in 10m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f878df2c46994f33a09625d53d44d327 : SUCCESS in 14m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41254ae0904844148cd0e3e7391115e1 : SUCCESS in 9m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/f8cf7c3057cf4fb6b5d7fe724d051db3 : SUCCESS in 52m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b384fb585d1341d5bb84a8864366aa37 : SUCCESS in 1h 09m 21s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a5780733605546e3ab7af88cc2c09ab4 : SUCCESS in 22m 55s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/c383bae32ced44e787a8fc6ad6c5df86 : SUCCESS in 27m 31s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6e7774f60524a5fb34c9aa4dc1ba844 : FAILURE in 20m 16s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/19ff696155b9456b9b0691eba2c002c4 : SUCCESS in 23m 47s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/5d568e16564d41bf927cf4ddd3834d1e : SUCCESS in 12m 15s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/147a1fb2ebd34f61ae88a29e93c95b0c : FAILURE in 14m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ee2b6e1277e468fb57bf348c1355160 : SUCCESS in 45m 19s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/39599f3d7081401982995b7dda5e785d : SUCCESS in 33m 34s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/ac527c099993497da90f9d84850471e0 : SUCCESS in 6m 04s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"73954ca48dd45e9e9a2dddbff61343282b8fcfb5","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-09-17 17:44:24.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2bf92d4c95eed9cfe9418536cb0a3557f04854c7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-17 18:11:31.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"5580d74d13186c8f9f9ca206defae2d1b17c534f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 20:30:41.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed84fff40d0044dfba82f4f8ad88d6a9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c586f4b2a43c437e92023786956bf051 : SUCCESS in 52m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4e8ea8d04f8442c1bf3456980c136e92 : SUCCESS in 24m 16s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"736824c8e0fbb85cacb31606eaf5843b8c67da91","author":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"date":"2024-11-01 19:52:04.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1f8b8e45fa2a893337ead2351321390fc49b17ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"date":"2024-11-01 19:52:52.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"baba166e1de8ef876d4ce0d83198fe5683af022c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"date":"2024-11-01 20:25:12.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"682ca13e27f55a4f96a8678b8a123e32b99724b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 21:44:31.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ad1ad7e16ae4b68b1dab4c9ead6285f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33e5839e54e14061b49a62a71313a800 : TIMED_OUT in 1h 00m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8629e93876444b8981ff17e44b626721 : TIMED_OUT in 1h 00m 48s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"83482ec502efb325e58c6d6082c98ef123ec1b7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 21:59:46.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2fb864a08494785a34e3cc33069c03a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38c6bec1316f48549d22d2a2048cddbc : SUCCESS in 14m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00040e7f3d714cb09548e903fb7ef15f : SUCCESS in 8m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2364f7c1b66d4c2a86aa3b8f833b6561 : SUCCESS in 12m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/77ab46f5e8eb449c809d1e8855232ee7 : SUCCESS in 10m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a2795589cce4e4c814df466a5748558 : SUCCESS in 12m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/c5fab5c494894543ad5b8c3284269417 : SUCCESS in 53m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15eae928e1654081869f923fa66a6433 : SUCCESS in 1h 26m 44s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/fe96942880dd4a16b38113e5870ed6fd : SUCCESS in 27m 17s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/38bb6cd3d5d74c22af7dda91bbe32584 : SUCCESS in 34m 54s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8909bd4412d4b6ea8397d4c99c73f25 : FAILURE in 16m 46s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/1784753cd43a4667abfa057ff48944e6 : SUCCESS in 17m 24s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/542d21f35aa04bb0bca3be6acd44e88e : SUCCESS in 13m 38s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/e20c65084eab4b6bb654d381d6ed45b2 : FAILURE in 12m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20f3686874a44ff4bf4bed821a6adab3 : SUCCESS in 39m 36s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/0c2569482e744abc957366f4fc2e263b : SUCCESS in 26m 12s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/9e48474a91fa460a8fded81f927de11b : SUCCESS in 4m 03s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"1caad6913f75a07e1be9b72e47c8d057d45dc601","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-02-14 14:47:06.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"242d8cda10e00cf509864f083a49cb83d432993c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 14:47:28.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56cd87aafa6743389f80228c83c6d9bc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1905ee91f0814ef181b5eab01e447cfb : NODE_FAILURE Node(set) request 300-0026339592 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b7fed31a99d249f6bdf38957c1d7eb0c : NODE_FAILURE Node(set) request 300-0026339593 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"fb58825c5d05793e1c90d9a31f17837dcd88b64a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 16:41:13.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4608e05e8e0a46a99fef52f3e8ab3196\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45f3b5c7c7414917a78ca73fa3f0e979 : SUCCESS in 15m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/261d4b96f2834396999cabf023a0aac5 : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b722e7d939a489698dad303778c8a5a : SUCCESS in 10m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a74b8796d0fe4fdfa6ab901c7724d3a6 : SUCCESS in 11m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aecceb08d9047ef889f59f1bb0f5f46 : SUCCESS in 13m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/082068486bf74c7fa7bbb2a6205a18d4 : SUCCESS in 1h 11m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89e2e6e810a54030abb8836f3c9f4c46 : SUCCESS in 1h 50m 52s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/aa31b29603fa434ea128ccec2e535d7a : SUCCESS in 29m 16s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/11ffaa1924954bc2886410c3e01f9af2 : SUCCESS in 47m 24s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/758a9a5d57624eb89f8f80bbc4724ac0 : FAILURE in 34m 06s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/28eb58e706284a98bebb8bc2694e43ad : SUCCESS in 29m 55s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/621939bb7802479781afb12ebd281317 : SUCCESS in 26m 36s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/f5ee1d5ddfe24dc287054430733c2063 : FAILURE in 12m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0dc1051dca9c41f3ab0d8fe589d983f0 : SUCCESS in 59m 10s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/a92255ea304241119c8acca1bb971229 : FAILURE in 42m 53s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/175b7b42d5664b29ad3e60abb5a3e67a : SUCCESS in 3m 59s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/cda666caf7834dccb88fb87c8b3ee046 : SUCCESS in 30m 13s","accounts_in_message":[],"_revision_number":12},{"id":"c4ddb28bac5ea1cbb6fd9624a86c3f3a8b8d155c","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-02-17 11:04:34.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"00ad5047293d2889214ef2e94494c1fd204cd5c8","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-02-21 14:09:54.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"28f6cd64428d63bd1bae0059b7cddda8ca5c1a69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 14:10:26.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"af6db3bfc9655fae40e458cc27e7c12bffdef646","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-21 14:20:09.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"01aca42a723eeb858542b490509584045dedad40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 16:32:17.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d08efe7a3fc2425d82bb103b39ed714a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b56d82e0e7bd446ca6688568b5bcdb8e : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90e761f7c9c849948e9b9264e5b079ed : SUCCESS in 13m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/26fdd32b8aab4477ab7eebdffb535f66 : SUCCESS in 13m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57741ee44470444fa5008009ba729527 : SUCCESS in 14m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/8ad37f9707384c6881770e26d1635ff9 : SUCCESS in 55m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d17ba868527b45b3a4e3917cbf98c0fe : SUCCESS in 1h 46m 46s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e1e94264885147d3880164017b957a61 : SUCCESS in 25m 51s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/5290344b7bb148449013df30d000b784 : SUCCESS in 31m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/773a2c6c620f4ed4ad2295eec3262e9b : SUCCESS in 51m 04s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/33d5f72886e0465ba62ade0e9d42e402 : FAILURE in 45m 41s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"50a1c5045814dc7bfda899d9136e1ddef1bd7067","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 16:32:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"42831eaa063fe6491bf9f9c8e5e297505a792e5c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 16:33:53.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d93bd8ee9e840b2b0cf3e287b42184d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f070b0de74d64e6bab1f48bc4feab890 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"1066c6c4bd55d8a692a23c07b4272f09a23da064","revisions":{"775ae4af21aa4779307cafd1b17fc314ad647821":{"kind":"REWORK","_number":1,"created":"2024-08-29 15:37:37.000000000","uploader":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"ref":"refs/changes/23/927523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/927523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/927523/1"}}},"commit":{"parents":[{"commit":"19b1f846b7927491ee6df262027f34a4b2224c70","subject":"Prevent circlular imports in keystone.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/19b1f846b7927491ee6df262027f34a4b2224c70"}]}],"author":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-08-29 15:36:00.000000000","tz":-420},"committer":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-08-29 15:36:00.000000000","tz":-420},"subject":"Add JSON Schema to `services` and validation decorators to services resource.","message":"Add JSON Schema to `services` and validation decorators to services resource.\n\nChange-Id: Iccef6011ff63f78d43a02fd2ccebbf6ac2522583\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/775ae4af21aa4779307cafd1b17fc314ad647821"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/775ae4af21aa4779307cafd1b17fc314ad647821"}]},"branch":"refs/heads/master"},"304a4083862d15bbf4181ef1c2484d03fdfe22d6":{"kind":"REWORK","_number":2,"created":"2024-08-29 17:24:37.000000000","uploader":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"ref":"refs/changes/23/927523/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/927523/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/927523/2"}}},"commit":{"parents":[{"commit":"19b1f846b7927491ee6df262027f34a4b2224c70","subject":"Prevent circlular imports in keystone.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/19b1f846b7927491ee6df262027f34a4b2224c70"}]}],"author":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-08-29 15:36:00.000000000","tz":-420},"committer":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-08-29 17:23:41.000000000","tz":-420},"subject":"Add JSON Schema to `services` and validation decorators to services resource.","message":"Add JSON Schema to `services` and validation decorators to services resource.\n\nChange-Id: Iccef6011ff63f78d43a02fd2ccebbf6ac2522583\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/304a4083862d15bbf4181ef1c2484d03fdfe22d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/304a4083862d15bbf4181ef1c2484d03fdfe22d6"}]},"branch":"refs/heads/master"},"b26261596252887e71f9e7d7fd66dc1799804fb3":{"kind":"REWORK","_number":3,"created":"2024-08-29 20:30:26.000000000","uploader":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"ref":"refs/changes/23/927523/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/927523/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/927523/3"}}},"commit":{"parents":[{"commit":"19b1f846b7927491ee6df262027f34a4b2224c70","subject":"Prevent circlular imports in keystone.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/19b1f846b7927491ee6df262027f34a4b2224c70"}]}],"author":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-08-29 15:36:00.000000000","tz":-420},"committer":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-08-29 20:28:06.000000000","tz":-420},"subject":"Add JSON Schema to `services` and validation decorators to services resource.","message":"Add JSON Schema to `services` and validation decorators to services resource.\n\nChange-Id: Iccef6011ff63f78d43a02fd2ccebbf6ac2522583\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b26261596252887e71f9e7d7fd66dc1799804fb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b26261596252887e71f9e7d7fd66dc1799804fb3"}]},"branch":"refs/heads/master"},"bae8c79bcc28cba4039dd5364b020aea8161132a":{"kind":"REWORK","_number":4,"created":"2024-08-30 16:03:07.000000000","uploader":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"ref":"refs/changes/23/927523/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/927523/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/927523/4"}}},"commit":{"parents":[{"commit":"19b1f846b7927491ee6df262027f34a4b2224c70","subject":"Prevent circlular imports in keystone.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/19b1f846b7927491ee6df262027f34a4b2224c70"}]}],"author":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-08-29 15:36:00.000000000","tz":-420},"committer":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-08-30 16:02:21.000000000","tz":-420},"subject":"Add JSON Schema to `services` and validation decorators to services resource.","message":"Add JSON Schema to `services` and validation decorators to services resource.\n\nChange-Id: Iccef6011ff63f78d43a02fd2ccebbf6ac2522583\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bae8c79bcc28cba4039dd5364b020aea8161132a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bae8c79bcc28cba4039dd5364b020aea8161132a"}]},"branch":"refs/heads/master"},"dcf1c6b45917564396e0babd49663b36038bb7e7":{"kind":"REWORK","_number":5,"created":"2024-08-30 16:51:20.000000000","uploader":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"ref":"refs/changes/23/927523/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/927523/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/927523/5"}}},"commit":{"parents":[{"commit":"b00cb4197d603da948f8f43f9c5f0e55182d2c96","subject":"Prevent circlular imports in keystone.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b00cb4197d603da948f8f43f9c5f0e55182d2c96"}]}],"author":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-08-29 15:36:00.000000000","tz":-420},"committer":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-08-30 16:51:04.000000000","tz":-420},"subject":"Add JSON Schema to `services` and validation decorators to services resource.","message":"Add JSON Schema to `services` and validation decorators to services resource.\n\nChange-Id: Iccef6011ff63f78d43a02fd2ccebbf6ac2522583\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/dcf1c6b45917564396e0babd49663b36038bb7e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/dcf1c6b45917564396e0babd49663b36038bb7e7"}]},"branch":"refs/heads/master"},"175defe90e9820d4f98e68ef67b006c8f0f616de":{"kind":"REWORK","_number":6,"created":"2024-09-03 16:11:08.000000000","uploader":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"ref":"refs/changes/23/927523/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/927523/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/927523/6"}}},"commit":{"parents":[{"commit":"b00cb4197d603da948f8f43f9c5f0e55182d2c96","subject":"Prevent circlular imports in keystone.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b00cb4197d603da948f8f43f9c5f0e55182d2c96"}]}],"author":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-08-29 15:36:00.000000000","tz":-420},"committer":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-09-03 16:10:27.000000000","tz":-420},"subject":"Add JSON Schema to `services` and validation decorators to services resource.","message":"Add JSON Schema to `services` and validation decorators to services resource.\n\nChange-Id: Iccef6011ff63f78d43a02fd2ccebbf6ac2522583\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/175defe90e9820d4f98e68ef67b006c8f0f616de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/175defe90e9820d4f98e68ef67b006c8f0f616de"}]},"branch":"refs/heads/master"},"ae204aad0fbea364efda263778005b4332d5b80b":{"kind":"REWORK","_number":7,"created":"2024-09-09 17:32:43.000000000","uploader":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"ref":"refs/changes/23/927523/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/927523/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/927523/7"}}},"commit":{"parents":[{"commit":"b00cb4197d603da948f8f43f9c5f0e55182d2c96","subject":"Prevent circlular imports in keystone.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b00cb4197d603da948f8f43f9c5f0e55182d2c96"}]}],"author":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-08-29 15:36:00.000000000","tz":-420},"committer":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-09-09 17:32:25.000000000","tz":-420},"subject":"Add JSON Schema to `services` and validation decorators to services resource.","message":"Add JSON Schema to `services` and validation decorators to services resource.\n\nChange-Id: Iccef6011ff63f78d43a02fd2ccebbf6ac2522583\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ae204aad0fbea364efda263778005b4332d5b80b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ae204aad0fbea364efda263778005b4332d5b80b"}]},"branch":"refs/heads/master"},"6a3306513df4f553f3f3441d86a3093b8a52ff09":{"kind":"REWORK","_number":8,"created":"2024-09-10 15:50:04.000000000","uploader":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"ref":"refs/changes/23/927523/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/927523/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/927523/8"}}},"commit":{"parents":[{"commit":"b00cb4197d603da948f8f43f9c5f0e55182d2c96","subject":"Prevent circlular imports in keystone.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b00cb4197d603da948f8f43f9c5f0e55182d2c96"}]}],"author":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-08-29 15:36:00.000000000","tz":-420},"committer":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-09-10 15:49:40.000000000","tz":-420},"subject":"Add JSON Schema to `services` and validation decorators to services resource.","message":"Add JSON Schema to `services` and validation decorators to services resource.\n\nChange-Id: Iccef6011ff63f78d43a02fd2ccebbf6ac2522583\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6a3306513df4f553f3f3441d86a3093b8a52ff09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6a3306513df4f553f3f3441d86a3093b8a52ff09"}]},"branch":"refs/heads/master"},"33715a12afa474e570f6150ce48e55a931c97df8":{"kind":"REWORK","_number":9,"created":"2024-09-17 16:29:31.000000000","uploader":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"ref":"refs/changes/23/927523/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/927523/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/927523/9"}}},"commit":{"parents":[{"commit":"b00cb4197d603da948f8f43f9c5f0e55182d2c96","subject":"Prevent circlular imports in keystone.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b00cb4197d603da948f8f43f9c5f0e55182d2c96"}]}],"author":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-08-29 15:36:00.000000000","tz":-420},"committer":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-09-17 16:28:37.000000000","tz":-420},"subject":"Add JSON Schema to `services` and validation decorators to services resource.","message":"Add JSON Schema to `services` and validation decorators to services resource.\n\nChange-Id: Iccef6011ff63f78d43a02fd2ccebbf6ac2522583\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/33715a12afa474e570f6150ce48e55a931c97df8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/33715a12afa474e570f6150ce48e55a931c97df8"}]},"branch":"refs/heads/master"},"cfd3944bb327c8bb1a6b652078af7b2844a41bb1":{"kind":"REWORK","_number":10,"created":"2024-11-01 19:52:52.000000000","uploader":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"ref":"refs/changes/23/927523/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/927523/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/927523/10"}}},"commit":{"parents":[{"commit":"b00cb4197d603da948f8f43f9c5f0e55182d2c96","subject":"Prevent circlular imports in keystone.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b00cb4197d603da948f8f43f9c5f0e55182d2c96"}]}],"author":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-08-29 15:36:00.000000000","tz":-420},"committer":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-11-01 19:52:13.000000000","tz":-420},"subject":"Add JSON Schema to `services` and validation decorators to services resource.","message":"Add JSON Schema to `services` and validation decorators to services resource.\n\nChange-Id: Iccef6011ff63f78d43a02fd2ccebbf6ac2522583\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/cfd3944bb327c8bb1a6b652078af7b2844a41bb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/cfd3944bb327c8bb1a6b652078af7b2844a41bb1"}]},"branch":"refs/heads/master"},"2a3b1a669c01b8b1da47c01ac6e5eb4da5f0cd6f":{"kind":"REWORK","_number":11,"created":"2024-11-01 20:25:12.000000000","uploader":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"ref":"refs/changes/23/927523/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/927523/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/927523/11"}}},"commit":{"parents":[{"commit":"b00cb4197d603da948f8f43f9c5f0e55182d2c96","subject":"Prevent circlular imports in keystone.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b00cb4197d603da948f8f43f9c5f0e55182d2c96"}]}],"author":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-08-29 15:36:00.000000000","tz":-420},"committer":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-11-01 20:24:10.000000000","tz":-420},"subject":"Add JSON Schema to `services` and validation decorators to services resource.","message":"Add JSON Schema to `services` and validation decorators to services resource.\n\nChange-Id: Iccef6011ff63f78d43a02fd2ccebbf6ac2522583\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2a3b1a669c01b8b1da47c01ac6e5eb4da5f0cd6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2a3b1a669c01b8b1da47c01ac6e5eb4da5f0cd6f"}]},"branch":"refs/heads/master"},"1066c6c4bd55d8a692a23c07b4272f09a23da064":{"kind":"REWORK","_number":12,"created":"2025-02-14 14:47:06.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/23/927523/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/927523/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/927523/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/927523/12"}}},"commit":{"parents":[{"commit":"20484381c71ed903c639386f2b73b031a11e334b","subject":"Merge \"Add JSON schema and validation for role grants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/20484381c71ed903c639386f2b73b031a11e334b"}]}],"author":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-08-29 15:36:00.000000000","tz":-420},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-02-14 14:46:53.000000000","tz":60},"subject":"Add JSON Schema to `services` and validation decorators to services resource.","message":"Add JSON Schema to `services` and validation decorators to services resource.\n\nChange-Id: Iccef6011ff63f78d43a02fd2ccebbf6ac2522583\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1066c6c4bd55d8a692a23c07b4272f09a23da064"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1066c6c4bd55d8a692a23c07b4272f09a23da064"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
