)]}'
{"id":"openstack%2Fkeystone~928923","triplet_id":"openstack%2Fkeystone~master~I739d28fff9304d58c4f278c5faecc4dddf755f9e","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-09-20 16:43:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I739d28fff9304d58c4f278c5faecc4dddf755f9e","subject":"Deprecate ineffective [DEFAULT] max_param_size","status":"MERGED","created":"2024-09-11 14:26:30.000000000","updated":"2024-09-20 16:47:01.000000000","submitted":"2024-09-20 16:43:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928923","meta_rev_id":"989f24301e9e1257534f26bdf922bb708ef7ce34","_number":928923,"virtual_id_number":928923,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-20 16:43:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2024-09-11 14:29:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-09-20 14:38:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"date":"2024-09-14 03:01:38.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-09-20 14:38:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-11 14:29:42.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-09-11 16:20:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-20 14:38:19.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"d89574b2f69a9307a20c159f58c2011d8a7783b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-11 14:26:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"514c21589db3379c9e2e67d8b295303a5202b9ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-11 14:28:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0273c3aac57580d66479d8e7a7fa71c6294eed21","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-09-11 14:29:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f363802fd1f218d265878aacf0c96236cde5247a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-11 15:21:11.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"71339588c6534a0075929338e76e5ae71f9c060d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 16:20:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c64ec7c4f2948e3b70c40e0eec3dc0e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1616632b98b4074b403efbd02a77bf6 : SUCCESS in 14m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71ddf411369540db9f7e7e83bda25238 : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87b6d69792c340548b878e7eb6bc3e9d : SUCCESS in 13m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a057e80435cb4fa69564c321d7ec8586 : SUCCESS in 10m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f7bc7657833a4836901156ac23e6f47f : SUCCESS in 12m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/324817305ae944f6810fdb0c747e402a : SUCCESS in 14m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/f944bde0a1b74e2eae4c4fb9beb4e5e8 : SUCCESS in 56m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/361d6c00df5c4d118e6c6d35e21937ac : SUCCESS in 1h 44m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07573a48b30742ea9fa4f80b3633cd5b : SUCCESS in 9m 38s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/14fc538ba597421d9eace88c7a1abb66 : SUCCESS in 31m 12s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/ab94de43424245d9897a890a9768fed4 : SUCCESS in 32m 28s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/718f7524af5b465bb9488403485a56e0 : FAILURE in 33m 22s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/8e300fdde3d6424bbcb02e640e525f5b : SUCCESS in 28m 58s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/241484a9ed85440ba088d1adf2ba6ae4 : SUCCESS in 13m 06s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/1c7b7a4871d44deeb7cbdf29fd7d0769 : FAILURE in 13m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/472bf1c7926b4869a4ad0b8ad5f0fab1 : SUCCESS in 1h 02m 33s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/74f791ee82bb48479f529015555e8c74 : SUCCESS in 44m 29s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/f8e8da4c03cd462b98118724098cd440 : SUCCESS in 6m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a534ef0ece7866822874208d777e5ab24d8b89a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 17:45:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efa5fb639c28464abb079fe3de31946d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/37e06c7953184bb99a2a3ed3a324ece5 : TIMED_OUT in 1h 00m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ec2539d6b6854b3cbf861933bfda51fe : TIMED_OUT in 1h 00m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9f15b6f8365d0100acfd0af4eda11e889f6f618c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-14 03:01:38.000000000","message":"Patch Set 2: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a008d48b1e3f3e730cf77a6e1bf349f468ae14d1","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-09-20 14:38:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7f473f35b0b999b856f8f1da408e8c1e947f9299","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 14:38:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d7f43a3320b1e75c91593b7c538ff1a3786adde1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 16:43:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e77bc1da6f394acd9cad0578c116065b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72e5164510f14d158f69d9a9decd9888 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17a6576ab4424147ad244f473251d22e : SUCCESS in 10m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7de6a9a78d124e2d86c71c8485614d22 : SUCCESS in 14m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21f8714c64c04c7490fe44be4c4b88fe : SUCCESS in 12m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/226a0e1a0340407f9296fc1ab47c7919 : SUCCESS in 54m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25e9b33f5daf4d20b6a98df952cc450e : SUCCESS in 1h 43m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1b53e19d82a490296f1d31f3cf6664b : SUCCESS in 10m 55s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/fc59e3744bca48589d7453d8155bd734 : SUCCESS in 26m 45s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/87a2e5711fcf41aca764307f56e30142 : SUCCESS in 31m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/005a2581847a4abd9793f1e47fec401f : SUCCESS in 1h 02m 05s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/53f8e87ef727411dadfa641dc00de77b : SUCCESS in 30m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c444f3575b7832b16d567ef336fff5eb2d2848fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 16:43:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"989f24301e9e1257534f26bdf922bb708ef7ce34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 16:47:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/511d54de84b24204bee5a59693f34098\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d0a2ffd975d4a729584c3b7b88aebbe : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc1a815c7b0541d09f55b3f8d307519c : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7ac41c21eaf2f618d447dfb4edfa0b6b37e7691e","revisions":{"9ba3847d05cecada60a1c5132ddad04bb5442b3d":{"kind":"REWORK","_number":1,"created":"2024-09-11 14:26:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/928923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/928923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/928923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/928923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/928923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/928923/1"}}},"commit":{"parents":[{"commit":"97431ec99b255f3e04d4b21d38412e5d6b798618","subject":"Merge \"Remove logic to support pysaml2\u003c3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/97431ec99b255f3e04d4b21d38412e5d6b798618"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-11 14:26:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-11 14:26:02.000000000","tz":540},"subject":"Deprecate ineffective [DEFAULT] max_param_size","message":"Deprecate ineffective [DEFAULT] max_param_size\n\nThis option is not used by any logic in keystone now.\n\nChange-Id: I739d28fff9304d58c4f278c5faecc4dddf755f9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9ba3847d05cecada60a1c5132ddad04bb5442b3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9ba3847d05cecada60a1c5132ddad04bb5442b3d"}]},"branch":"refs/heads/master"},"7ac41c21eaf2f618d447dfb4edfa0b6b37e7691e":{"kind":"REWORK","_number":2,"created":"2024-09-11 14:28:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/928923/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/928923/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/928923/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/928923/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/928923/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/928923/2"}}},"commit":{"parents":[{"commit":"97431ec99b255f3e04d4b21d38412e5d6b798618","subject":"Merge \"Remove logic to support pysaml2\u003c3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/97431ec99b255f3e04d4b21d38412e5d6b798618"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-11 14:26:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-11 14:28:44.000000000","tz":540},"subject":"Deprecate ineffective [DEFAULT] max_param_size","message":"Deprecate ineffective [DEFAULT] max_param_size\n\nThis option has been unused since identity v2 API was removed[1].\n\n[1] d5e9c0b4fe814ba6e6d36a7d1ea676914c79e5eb\n\nChange-Id: I739d28fff9304d58c4f278c5faecc4dddf755f9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7ac41c21eaf2f618d447dfb4edfa0b6b37e7691e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7ac41c21eaf2f618d447dfb4edfa0b6b37e7691e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
